)]}'
{"id":"openstack%2Fzun~867086","triplet_id":"openstack%2Fzun~master~I5dd1e4057722ddece259c70985b77a3dd1dfd48b","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-12-11 03:44:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}},"hashtags":[],"change_id":"I5dd1e4057722ddece259c70985b77a3dd1dfd48b","subject":"Disable git owner check for zun","status":"MERGED","created":"2022-12-09 08:57:02.000000000","updated":"2022-12-11 04:03:36.000000000","submitted":"2022-12-11 04:02:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867086","meta_rev_id":"bc51d74f16207f52b741eb99e33aa79677891919","_number":867086,"virtual_id_number":867086,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-11 04:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-12-11 03:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-12-11 03:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 08:57:11.000000000","updated_by":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-09 09:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68c50f7fd16b06dfd1950c49ae35bfb1176aae51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-09 08:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"024f1915ff1fae7f045d15907cd69e6b73fbafd5","author":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"date":"2022-12-09 08:57:11.000000000","message":"Patch Set 1:\n\nBuild Started https://build.opnfv.org/ci/job/releng-tox/5/","accounts_in_message":[],"_revision_number":1},{"id":"a72d3452005bbfe232bf6065f181055aaf490d34","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-09 09:06:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ffaed92bcab358e4d03de28ea59a06af1793c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 09:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db3b2b62e0534013837872b4891979bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49c6bb2ff28b4557be6c3abdff0d86a3 : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8904b8775d4446758ca01cff286047bb : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1343be84c6574cf7075a196850e90b26d85bb3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-11 03:43:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"74a410b547cbac6022cdfb60bd31488845bb5b77","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-11 03:44:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b1c97e795d69221ddaae25476409b571a42b46c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 03:50:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94bbd20305ab4f7ea00d8bd6d5772ed6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218ef0cbcf5440f7a7282d5f1569d8b3 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74e19e592c64822b3c410a9b5ed7fba : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"05ab0b694d5fafa86edce86332bc4fcd67b8a30e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 03:50:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1210267b972c6d5df67d2c45c4c5acf218d4b4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 04:02:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33667b508e4c43d88eb95ce1a71be79b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5acd057d61e4845b248893ebeedd846 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4acede02ad9449fe8395e8ebf75fa42c : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"86655d192d2e8ab762cf209cc79a4f439573cc97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 04:02:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bc51d74f16207f52b741eb99e33aa79677891919","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 04:03:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd3fc59c62e4858924453fb6847ef92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d168337d36435a969a111401e2b4cb : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8244086d1fec030b10dbf2e066859ce03761f0fd","revisions":{"96dbf62a274792ff81cd05feb9eb5570b5db525d":{"kind":"REWORK","_number":1,"created":"2022-12-09 08:57:02.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/86/867086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/86/867086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/86/867086/1"}}},"commit":{"parents":[{"commit":"f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-12-09 08:52:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-12-09 08:52:52.000000000","tz":0},"subject":"Disable git owner check for zun","message":"Disable git owner check for zun\n\nThe new git owner check [1] broke zun installation process.\nWe add an exception to skip the check.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n\nChange-Id: I5dd1e4057722ddece259c70985b77a3dd1dfd48b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/96dbf62a274792ff81cd05feb9eb5570b5db525d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/96dbf62a274792ff81cd05feb9eb5570b5db525d"}]},"branch":"refs/heads/master"},"8244086d1fec030b10dbf2e066859ce03761f0fd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-11 03:43:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/86/867086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/86/867086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/867086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/86/867086/2"}}},"commit":{"parents":[{"commit":"7bddd4f42c4a50939331ac55768f1825f23b9022","subject":"Handle io_service_bytes_recursive is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7bddd4f42c4a50939331ac55768f1825f23b9022"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-12-09 08:52:52.000000000","tz":0},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2022-12-11 03:43:39.000000000","tz":0},"subject":"Disable git owner check for zun","message":"Disable git owner check for zun\n\nThe new git owner check [1] broke zun installation process.\nWe add an exception to skip the check.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n\nChange-Id: I5dd1e4057722ddece259c70985b77a3dd1dfd48b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8244086d1fec030b10dbf2e066859ce03761f0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8244086d1fec030b10dbf2e066859ce03761f0fd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[]}
