)]}'
{"id":"openstack%2Fzun~867110","triplet_id":"openstack%2Fzun~master~I925e3b44cd321ef593cf59a08dee2c428c6bee96","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-12-10 15:21:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}},"hashtags":[],"change_id":"I925e3b44cd321ef593cf59a08dee2c428c6bee96","subject":"Also add B324 to bandit skip list for python3.9+","status":"ABANDONED","created":"2022-12-09 14:04:43.000000000","updated":"2022-12-10 15:24:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0d2550716bbe88b451422f4b00933a3a94cfdb7","_number":867110,"virtual_id_number":867110,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-09 14:40:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2022-12-10 15:21:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2022-12-10 15:21:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 14:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc20365d1c30471df3ae3390aff597979ff56f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-09 14:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3309fa75fab4b6476fe516b9e968ecf493b49873","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-09 14:24:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fbb6ff7a1213ba11e73467e55fe83db8c75208b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 14:40:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57e198489a7340849ce20c54126ca094\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f8da5105d574534b5e3e18026cb5aca : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af2c286a4654d04902772759883018d : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb0af8496074e4e874a4aed006f1489 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eafb173e8bc0430aa8e882d52712da6d : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bd0b357de9340c0933086fdca775609 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b71c5cfc6d47cab6ef224e92d4a00d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eb1c05222eb4d06922e00e72f91a580 : SUCCESS in 3m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9dc832003dde490d8ed8edae8f385dd7 : FAILURE in 32m 00s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/01393512a76b431182f99485651c331a : SUCCESS in 34m 13s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e2009eb5d01140f081648ccefc051e69 : FAILURE in 24m 36s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/3d08aedbe9d04596b73d5a3d83ca9323 : FAILURE in 23m 17s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2e2bb247ae014f599b61823a87403152 : FAILURE in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9217d8f9f12cd3b78c367cfb207abad80b531244","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-10 06:17:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a0e3507b8eeb87490be1ef9eaa09923b341d706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 07:08:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5d065aca85e42c9a094604f9c310354\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/74a1bc1358ea4dcea773293f9763ca78 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f253050fdaf4224a8df10f8145116f7 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024e543a6330421fa88af89dd565b82f : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eb72940558b4b8d8e67f28f27d18816 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ff1e559aab943d19714a0c95a563df1 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c34e08a7ed4a82a2bd5b96a7005d80 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/809e18dd3f0343f8a3f9a108e6954586 : SUCCESS in 3m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/6ad8283e4a804ea7a581d3bcaaffb9c2 : FAILURE in 31m 17s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/a0e0ea1e9a9f4ade9ff3bf7ddbcc8e52 : SUCCESS in 37m 34s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7a53272f4c104c538e691c6dab811e9e : SUCCESS in 41m 34s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/1afa69fcc7144429a51456ad0058218e : SUCCESS in 24m 58s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3512df3e270846dbbc656a4bca2098d5 : FAILURE in 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8bb56f670bfc29ff03603fc415ce844f3182c8a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-10 15:21:04.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a0d2550716bbe88b451422f4b00933a3a94cfdb7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-12-10 15:24:18.000000000","message":"Abandoned\n\nFixed in https://review.opendev.org/c/openstack/zun/+/867173","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a2b9d7ca244962d0cb92ed7564dfb8e2d999fa8","revisions":{"1a2b9d7ca244962d0cb92ed7564dfb8e2d999fa8":{"kind":"REWORK","_number":1,"created":"2022-12-09 14:04:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/10/867110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/10/867110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/10/867110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/10/867110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/10/867110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/10/867110/1"}}},"commit":{"parents":[{"commit":"f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f7f67cbf758e8c0b7dd9387b1607fa2f3d745c78"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-12-09 14:04:28.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2022-12-09 14:04:43.000000000","tz":0},"subject":"Also add B324 to bandit skip list for python3.9+","message":"Also add B324 to bandit skip list for python3.9+\n\nChange-Id: I925e3b44cd321ef593cf59a08dee2c428c6bee96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1a2b9d7ca244962d0cb92ed7564dfb8e2d999fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1a2b9d7ca244962d0cb92ed7564dfb8e2d999fa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
