)]}'
{"id":"openstack%2Fzun~872679","triplet_id":"openstack%2Fzun~master~Ied83faa9864eaaaadc05cbfaae7a082f09ac9464","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ied83faa9864eaaaadc05cbfaae7a082f09ac9464","subject":"modify","status":"ABANDONED","created":"2023-02-06 07:50:34.000000000","updated":"2026-08-29 19:58:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"5592f34bfbc090d18108c102e76108cb01a934e3","_number":872679,"virtual_id_number":872679,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-06 08:27:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-02-06 08:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf74aeefca2b831603fab47fbd597963524af368","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2023-02-06 07:50:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33c6ca3e1db8ffd9c50f0bef762eefb00a5b19a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 08:27:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dd7a773a02d44a29a5a45618d3858e5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/707111c90c214dc4b3d55de49b9d3446 : SUCCESS in 4m 02s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/46720e3e6a6a4d26a7f43ec66b1baebb : SUCCESS in 1m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e92a4b754014205b602e5a078f350e6 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e1b1f63cca64509a1758a27cc8f22b4 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e5bf1d316b487485443b3345cd2792 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5db6f68ef5648fd86ea23972c0947ec : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/297564655b134492a6be452e2fe79089 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed67c2ca210e414b99268a3a27498e21 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c33c971e6c464fe3818aee0af13c01d5 : SUCCESS in 3m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7b1b063b239043fdba2e3cf745358d13 : FAILURE in 23m 04s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9088b02633334197ade8431ffc10ad45 : FAILURE in 30m 55s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5d768f365b7c4f19b8cd888a3ad1afc1 : FAILURE in 26m 45s\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/88de1a9cbc7849488b035968a4b400bc : FAILURE in 17m 52s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8e453d836f104a4fadf3a9c3a6601c87 : FAILURE in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5592f34bfbc090d18108c102e76108cb01a934e3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 19:58:26.000000000","message":"Abandoned\n\nHello, MR without any changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"587e7bb2435bead939250771672ff77643e0fdec","revisions":{"587e7bb2435bead939250771672ff77643e0fdec":{"kind":"REWORK","_number":1,"created":"2023-02-06 07:50:34.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/79/872679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/79/872679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/872679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/872679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/872679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/79/872679/1"}}},"commit":{"parents":[{"commit":"cc9b8b3d2d14e58d32797ffed219f70bc9f77573","subject":"Use new get_rpc_client API from oslo.messaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/cc9b8b3d2d14e58d32797ffed219f70bc9f77573"}]}],"author":{"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","date":"2023-02-06 07:50:34.000000000","tz":0},"committer":{"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","date":"2023-02-06 07:50:34.000000000","tz":0},"subject":"modify","message":"modify\n\nChange-Id: Ied83faa9864eaaaadc05cbfaae7a082f09ac9464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/587e7bb2435bead939250771672ff77643e0fdec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/587e7bb2435bead939250771672ff77643e0fdec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
