)]}'
{"id":"openstack%2Fzun~935249","triplet_id":"openstack%2Fzun~master~I3c1f00293ad0b21de0d3692c81895a6a36f84f25","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-06 10:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c1f00293ad0b21de0d3692c81895a6a36f84f25","subject":"Replace deprecated configure_auth_token_middleware","status":"MERGED","created":"2024-11-15 02:10:42.000000000","updated":"2024-12-06 11:00:02.000000000","submitted":"2024-12-06 10:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935249","meta_rev_id":"85e2806350affa1153eac18e3812b397a65bbf5d","_number":935249,"virtual_id_number":935249,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 10:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2024-12-06 10:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2024-12-06 10:07:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 03:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 08:51:19.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2024-12-06 10:07:38.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"378cd944d1e219e0e8d0dbb82cf6d2c8e9f8e262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 02:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d62534afaae0a8dbdc4ebbf36a57aed0876ff88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 03:33:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7db2574efec44e6e8c9e555a500666e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa718209bdfa47d694113272111fd0cf : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a024f63a12a46f3953ee193bfa7d972 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13328d3986bf4e3c8ccaf2b011fde689 : SUCCESS in 7m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3524200563134df297a8607b2aa4c89c : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3cc906752a42fab72d44dca1de7dde : SUCCESS in 6m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9407a42dde024d8b8053a6e825d860ae : FAILURE in 32m 54s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/0bddb10b310e40c8969c8ad97df28d5c : FAILURE in 27m 09s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5d8068b6fde241a384b18f3f5aaa082f : FAILURE in 31m 23s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/50d66346ef844de8a9c774c9c9ab4b6f : FAILURE in 15m 33s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/935f74c2bc4741ce9af3beceb2088137 : FAILURE in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f18ca1b08deec3cfcc0f9f097749685806cbf90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 08:09:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"272ee2ac8b6a362a7516c8590d87a99286c94344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 08:48:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff1dada0bde446bfbcdfcb3e3618231f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a5e6f17f10b40f8914aa1b65992977a : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc768281e534c27a13420daf3a662de : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a651b3013ff84c8eb2b5d3265132eeab : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec5ac6ab5d03427ba03c28b524f5092f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fab2109f4249fc9fb2a444c51f32e3 : SUCCESS in 7m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/55fa1a71dfca4ac3a0182a8c3178d484 : FAILURE in 6m 42s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/abb7fbeb9baa498aac443adb554ed1c0 : FAILURE in 20m 45s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/959e8939f42e426d9442be7a90c4a2eb : FAILURE in 17m 04s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7c906fc71c7f43b1801aabc6d237c67b : FAILURE in 6m 38s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/e20cf81652554f9ea79931e3df47b49b : FAILURE in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9a7e8c4c8f246be2e27d332eb4a6839d38282f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 15:26:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"69d5b67ab4ac30bab6df326496e5643bfcfea586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:55:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a51f34a73f8c4bfb99ea63512ac4ef90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89c1ecc263a4002a36e76f2c053d3cf : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c69dfc48f0944b1a47cab3e69a11184 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c91e9eaddfe34d9cb23e1b2f640130fa : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc98489deeba4e99a71e0bfb200f93d0 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db8407b91b24c6f837beda647c3fecf : SUCCESS in 6m 11s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2c59379961be4d1e82e872e89cfccecd : FAILURE in 21m 56s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/7c9289b2f27e4a678248ab4cc6936265 : FAILURE in 9m 37s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/08d01ee1f5f640f48c0621205ab7e73a : FAILURE in 9m 05s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ca925b50a56d4965924de993a5979c90 : FAILURE in 7m 40s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/37bc912e1600447ea0f539670bd46c5e : FAILURE in 7m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8dd2282497655c3d08e842d4b8a4743484157026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 16:24:55.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ab7b763322e704b687c00caeb95789a51479a70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:22:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/640d2235a25e436dbb885c728dde13aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d620b94f667a45068a789eff272231fb : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd26331dd764c589502a34062d8b370 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f6087bea6914555bb5ee342867c1590 : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4b4eadea0ba4f04b13f1c908741d8d9 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81e58acf9a2491b94f5b5866c6c5d93 : SUCCESS in 7m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5ca0e8e25a314a61a5be79b6a2973d9a : FAILURE in 41m 06s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e577715d9e934e51a8d20713b307ea99 : FAILURE in 50m 13s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8fa7892ffede4927b97afb7dade012fd : FAILURE in 50m 05s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/3bbf28ba3cd14b769c0f6c04ce28034a : FAILURE in 42m 39s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/8cea98d77e9f432fae21a1ad7789dd70 : FAILURE in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fca0be9ce09b0c1497d221f9b29fdb27eae83e0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-06 08:51:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc947ba225608f7f9c032c5771f2345c8ae82441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 09:48:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649ff104562340eca05b23e6703d80d7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52bbfb76e33543078dda192bbb001d63 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abd5df7eea94612adb4af2437637109 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b99c380b2344368819a22e413eb8ff : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e90e6e46c74d416681ede578c648bd3b : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04b6285dbcc42248745ee0c5f3f518d : SUCCESS in 10m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a85e0cbfe23543e9a9bd7064b3a59f4f : SUCCESS in 51m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/1c5a5d07dfff4851aac8f5410c5d115f : SUCCESS in 43m 49s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/75dff6a265de45698f5f4ec9d802bf35 : FAILURE in 56m 54s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/7f56e17141bb495bafd951e888c11112 : FAILURE in 36m 36s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/fb561db4496d45debaf4b7c38c548a2a : FAILURE in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb9c4a59df07c49058d94c1d59cfdc531334db33","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-06 10:07:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8dd08eead8126425bba199dd04224358a5aea00f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 10:08:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"04f073cfd6f639bf4b1553a8da7bc468528a54da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 10:59:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd12365d55c4678914b4ad9d4cc55d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e7266f964d42e2802e95f281206c67 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29b5abee244a4b50aeb27783df79a241 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cb2825940a14107be6e23557fe42671 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fff2691ddd406f95ac6c1f5889019c : SUCCESS in 7m 21s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7e3e497dd4f04275b2935549a8b019e8 : SUCCESS in 45m 57s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ff4216f4e4e4c1ca9df712a1fd8f6d1 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":4},{"id":"0628539ee92cbff0750fb3d47c5e11acc9007e68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 10:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"85e2806350affa1153eac18e3812b397a65bbf5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 11:00:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e2eb0d3ff724c73bfac20370a70e2c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a12e285b16e489ea2e8b35d9fcc7bc0 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8a8b7b6acaec93050064f9ea25236ba50eac2b68","revisions":{"abb5b1e4b3967c45c41b37ed9ec50609898bad32":{"kind":"REWORK","_number":1,"created":"2024-11-15 02:10:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/935249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/935249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/935249/1"}}},"commit":{"parents":[{"commit":"5c5e1063edb452418cfe5b9dc452016171738312","subject":"Remove workaround for eventlet bug #592","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c5e1063edb452418cfe5b9dc452016171738312"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:10:40.000000000","tz":540},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] https://review.opendev.org/628651\n\nChange-Id: I3c1f00293ad0b21de0d3692c81895a6a36f84f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/abb5b1e4b3967c45c41b37ed9ec50609898bad32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/abb5b1e4b3967c45c41b37ed9ec50609898bad32"}]},"branch":"refs/heads/master"},"6d89cf0ada487b651eec13ceb7ce21707701c494":{"kind":"REWORK","_number":2,"created":"2024-11-15 08:09:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/935249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/935249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/935249/2"}}},"commit":{"parents":[{"commit":"5c5e1063edb452418cfe5b9dc452016171738312","subject":"Remove workaround for eventlet bug #592","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c5e1063edb452418cfe5b9dc452016171738312"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 08:09:23.000000000","tz":540},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] https://review.opendev.org/628651\n\nChange-Id: I3c1f00293ad0b21de0d3692c81895a6a36f84f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6d89cf0ada487b651eec13ceb7ce21707701c494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6d89cf0ada487b651eec13ceb7ce21707701c494"}]},"branch":"refs/heads/master"},"a7b1eb587b1800db9a8c286f3bb0a91c8f71ce71":{"kind":"REWORK","_number":3,"created":"2024-11-15 15:26:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/935249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/935249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/935249/3"}}},"commit":{"parents":[{"commit":"5c5e1063edb452418cfe5b9dc452016171738312","subject":"Remove workaround for eventlet bug #592","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c5e1063edb452418cfe5b9dc452016171738312"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 15:26:51.000000000","tz":0},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] https://review.opendev.org/628651\n\nChange-Id: I3c1f00293ad0b21de0d3692c81895a6a36f84f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a7b1eb587b1800db9a8c286f3bb0a91c8f71ce71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a7b1eb587b1800db9a8c286f3bb0a91c8f71ce71"}]},"branch":"refs/heads/master"},"8a8b7b6acaec93050064f9ea25236ba50eac2b68":{"kind":"REWORK","_number":4,"created":"2024-11-15 16:24:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/935249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/935249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/935249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/935249/4"}}},"commit":{"parents":[{"commit":"5c5e1063edb452418cfe5b9dc452016171738312","subject":"Remove workaround for eventlet bug #592","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c5e1063edb452418cfe5b9dc452016171738312"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 16:24:55.000000000","tz":0},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] https://review.opendev.org/628651\n\nChange-Id: I3c1f00293ad0b21de0d3692c81895a6a36f84f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8a8b7b6acaec93050064f9ea25236ba50eac2b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8a8b7b6acaec93050064f9ea25236ba50eac2b68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
