)]}'
{"id":"openstack%2Fzun~942801","triplet_id":"openstack%2Fzun~stable%2F2024.1~Ieb1f4d484ead3bf77293499fa931b5e2772d6f39","project":"openstack/zun","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"35987":{"account":{"_account_id":35987,"name":"Lana Kaleif","email":"lkaleif@mail.ru","username":"lkaleif"},"last_update":"2025-10-31 13:57:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ieb1f4d484ead3bf77293499fa931b5e2772d6f39","subject":"Remove pytz","status":"ABANDONED","created":"2025-02-26 14:05:35.000000000","updated":"2025-10-31 13:57:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":941075,"cherry_pick_of_patch_set":1,"meta_rev_id":"bd28905221c03ed0579dda128d507a63f20b8e8a","_number":942801,"virtual_id_number":942801,"owner":{"_account_id":35987,"name":"Lana Kaleif","email":"lkaleif@mail.ru","username":"lkaleif"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-26 15:26:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 14:05:35.000000000","updated_by":{"_account_id":35987,"name":"Lana Kaleif","email":"lkaleif@mail.ru","username":"lkaleif"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2025-02-26 15:26:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-31 13:57:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"c126fafc38fe3d8e959b57ae41f8bfc053db9be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35987,"name":"Lana Kaleif","email":"lkaleif@mail.ru","username":"lkaleif"},"date":"2025-02-26 14:05:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"48efd193b9a9974e15ece482fa54aaf3295968ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 15:26:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d055c5fd505f40519114968fc4a6fa29\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eef4b10bdb04d44a628d64f506458ba : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3f7a6fcd2f436c883727d2d06d667d : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5de0fee4e31439ab51d1895c48b624a : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af7a3b7ee9a94e8581d998057faab8f5 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3006b79efc904998b5ed8115bf0795a4 : SUCCESS in 6m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ef5842969ce34ff48535371b4785ea16 : TIMED_OUT in 1h 12m 11s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/84548f4ba94343d0a292b8ec31dbe2b4 : SUCCESS in 41m 23s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/041136000f4e415dbfbe610fbf3e5c83 : SUCCESS in 44m 05s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/ce132923aec74a1ab8e5bc64e8b9b948 : FAILURE in 44m 43s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/45d380299df8414bb881434c52418650 : SUCCESS in 1h 08m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"360e920b1880109b1af29e268f888854df7d4b8b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-02-27 04:12:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"374d1a860dbc1970945ffdf289e484c8eeb686a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 05:12:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1dd7d3cb2444bb1b788ea2c0aa88831\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9af585a8d93454f88da9dc874d9c4f7 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea901b0da3314f629af3b661bd4c6119 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ee2ac4e12694e49bf045445fa80d95b : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64e7929db5954cc4b087b79874d4f12f : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f9e7a205b1409b8161cf473144ef54 : SUCCESS in 6m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ae5ba58652db4e439275ce4cd7905796 : FAILURE in 55m 22s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/aabef65fb40e41468a4f688538ad8aa4 : SUCCESS in 43m 34s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b099633b0b1e43f481c16c1c60f1a9ca : SUCCESS in 48m 04s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/62030b6c850e4401932881cb80523285 : FAILURE in 37m 15s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/23c7e4c01516460d8855df71fc0275d7 : FAILURE in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd28905221c03ed0579dda128d507a63f20b8e8a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-10-31 13:57:09.000000000","message":"Abandoned\n\nstable/2024.1 branch of openstack/zun is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/2024.1 if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68cb3125e4fcde2597afd5318e1234a39326d396","revisions":{"68cb3125e4fcde2597afd5318e1234a39326d396":{"kind":"REWORK","_number":1,"created":"2025-02-26 14:05:35.000000000","uploader":{"_account_id":35987,"name":"Lana Kaleif","email":"lkaleif@mail.ru","username":"lkaleif"},"ref":"refs/changes/01/942801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/01/942801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/942801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/942801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/01/942801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/01/942801/1"}}},"commit":{"parents":[{"commit":"75d13bae26e9b9658277ffeba62e3b65767bb248","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2024.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75d13bae26e9b9658277ffeba62e3b65767bb248"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2025-02-10 06:03:44.000000000","tz":0},"committer":{"name":"Lana Kaleif","email":"lkaleif@mail.ru","date":"2025-02-26 14:05:35.000000000","tz":0},"subject":"Remove pytz","message":"Remove pytz\n\nChange-Id: Ieb1f4d484ead3bf77293499fa931b5e2772d6f39\n(cherry picked from commit 08ede38ca02b00d1639823498d9355d72db88d73)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/68cb3125e4fcde2597afd5318e1234a39326d396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/68cb3125e4fcde2597afd5318e1234a39326d396"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
