)]}'
{"id":"openstack%2Fzun~943094","triplet_id":"openstack%2Fzun~stable%2F2024.1~I5cb0df5c0c65a4e54841ffb9e8450b21446a5fdd","project":"openstack/zun","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2025-03-03 05:45:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}},"hashtags":[],"change_id":"I5cb0df5c0c65a4e54841ffb9e8450b21446a5fdd","subject":"Fix the installation document about removal of --cluster-store","status":"MERGED","created":"2025-03-01 11:00:53.000000000","updated":"2025-03-03 06:09:57.000000000","submitted":"2025-03-03 06:07:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943094","cherry_pick_of_change":942989,"cherry_pick_of_patch_set":1,"meta_rev_id":"80d8a78498b35c8ccf8288fc705e5ed146491baa","_number":943094,"virtual_id_number":943094,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 06:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2025-03-02 14:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2025-03-02 14:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 11:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25cbdd0b947656179df9011934eb6e22b5a4062e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-03-01 11:00:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"158c8c659cd48abcb4ddee236853308cd6b79555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 11:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae68fbf4d5084247abedeef7cba3f288\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1330a32b504259930cb9f467dce770 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353f70686a2d4146a9c4c223dfb07e8e : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"80691aadcf1f644b4ba924a546e4c0ea4cabb29d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-03-02 14:29:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b95b787c64e3a3cf3e02f250db104806334b9211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 14:29:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3a7e88249737c6d891e1288946794e9a5d31b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 14:49:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4500dadd9e3d4b5bbf1b156b9690cb3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37b990a68864758809083a7f36400ee : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4a4630b31042799d0ebb7428cc4103 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0d77270224c830b2b38d3794292a7170110e3847","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-03-03 05:45:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d50bbfe65fcef47a259af2f3b02a6ec960c5e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 05:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecb9ae42a1640b89dacf7a191e27f78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa018142a01d44b6ba1cfe55a4f8d25f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c929ca72b364a7c9652b322ce55009a : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2a2cb18d1121ad3a32e5fa408a54b8a88e3f0b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 05:58:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de6f953e23ede7e50362562944a4b8d8eb1e96c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 06:07:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c225faa31404592ad24369e2be4363b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23356c5f4754ab9800d9ebb33ecb6c0 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9025c0870c054324a28e57015f2270d9 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"81c170e8264a1961ce459575679a27c93e19a829","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 06:07:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80d8a78498b35c8ccf8288fc705e5ed146491baa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 06:09:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f84adb2178d4a58a65a70b739579b0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02076d32823744ef9f4ffb57b1c08d3e : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82be4a95d0a4ffc9debd390a0a71faa222ac185e","revisions":{"82be4a95d0a4ffc9debd390a0a71faa222ac185e":{"kind":"REWORK","_number":1,"created":"2025-03-01 11:00:53.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/94/943094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/94/943094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/943094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/943094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/94/943094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/94/943094/1"}}},"commit":{"parents":[{"commit":"75d13bae26e9b9658277ffeba62e3b65767bb248","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2024.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/75d13bae26e9b9658277ffeba62e3b65767bb248"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2025-02-28 10:30:11.000000000","tz":0},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2025-03-01 11:00:53.000000000","tz":0},"subject":"Fix the installation document about removal of --cluster-store","message":"Fix the installation document about removal of --cluster-store\n\nStarting from stable/2024.1, we no longer need a cluster store\nfor docker. Therefore, we change the installation steps as follows:\n* Remove the --cluster-store option from Docker\n* Remove the etcd requirement (since we don\u0027t need to use it as cluster store).\n* Switch capability_scope to \"local\" (the default) in kuryr-libnetwork.\n\nRelated-Bug: #2007142\nChange-Id: I5cb0df5c0c65a4e54841ffb9e8450b21446a5fdd\n(cherry picked from commit b480d7106e54f577d4d5819cbbce157f86f94fdb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/82be4a95d0a4ffc9debd390a0a71faa222ac185e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/82be4a95d0a4ffc9debd390a0a71faa222ac185e"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
