)]}'
{"id":"openstack%2Fzun~953725","triplet_id":"openstack%2Fzun~stable%2F2025.1~Ibd4bc5214c55f1379738a8df0c2e3a39a6e90df0","project":"openstack/zun","branch":"stable/2025.1","topic":"remove-wsgi_scripts-stable/2025.1","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2025-06-30 13:51:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}},"hashtags":[],"change_id":"Ibd4bc5214c55f1379738a8df0c2e3a39a6e90df0","subject":"Migrate from wsgi scripts to module paths","status":"MERGED","created":"2025-06-30 06:10:53.000000000","updated":"2025-06-30 15:31:03.000000000","submitted":"2025-06-30 15:30:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953725-remove-wsgi_scripts-stable/2025.1","cherry_pick_of_change":953551,"cherry_pick_of_patch_set":1,"meta_rev_id":"f5341c82572844b861c9617f9fd7301d4a56c64c","_number":953725,"virtual_id_number":953725,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 15:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2025-06-30 06:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2025-06-30 06:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 08:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22520cda97f6f67e91055b0f305b73e775773f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-30 06:10:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"80c1783121c9259e1da68e98577bc52f4742bcdf","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-30 06:11:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3368911e1cad40ae6b55605cdaf82a53615d5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 08:18:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ecabbb8b5644533a6d8bb2a4a488bdc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb5b4e717a944f6e8d1e80d48df76271 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c103fa482b3f4de68ca978e317d23f0f : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e61931b9ac484050b0049d02b1d2b982 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f133c7e8a29e4b4ea3ef68ecc0a6e0d7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49911246e6de418ea4a72e90133326c8 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46e456ab0c5d4a8684e16932942a5808 : SUCCESS in 2m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1df74028cf0a49488f4f4eef01eb7d40 : FAILURE in 52m 37s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd62a2663e164463a55b56a9b53c563f : SUCCESS in 20m 53s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e504a822f977498c955bebe9442f891e : SUCCESS in 21m 49s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d8d5f2aa30c3407a8ce865953b0013d8 : FAILURE in 2h 02m 07s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d935da1346d44e7aae6d85e7a6ab0e53 : FAILURE in 37m 33s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/74d3c8eb34fb486cb6fe1105e9d8093a : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d29740b5a6bb0e47f078af708c46b1b16e1348cc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-30 13:51:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ac7d1e5eb6451dceb08f1a481d1da31a4830a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:35:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd4cada337942ada36d71c08cd970be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48e97fcaba9544f98e7491ba25963076 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be1ff4a825744ea88ea7b590ac2964f : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3c0ecf523254d999d0c1e20110fbc6d : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2553ef39cde64db69a0f4f2c82ccb9e1 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0fca701b4a4e2dbdad78ae43e39eef : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/280b0c7a8de543e598d2eef52718c929 : SUCCESS in 2m 13s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5e554fa9679044a1a4fd9babf74cfc99 : SUCCESS in 23m 40s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/309a9c64b1af455fb7da8f6cf7fea29f : SUCCESS in 35m 03s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/5070e71748414b9d95ee1a076bd8db18 : SUCCESS in 37m 18s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b55b5034ef064395b93ce431f7778925 : SUCCESS in 38m 46s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/1b4201cdde1a4160a50503caac9152c5 : FAILURE in 32m 02s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/897c4bc9665045cc873aea0fe5cd15fc : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da4f4228534b464333e3fcdd9ca319967288643c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:36:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcf8ec3fa95b2c9bbe03f2d23f07258de47591f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 15:30:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b40fd49d9144d2bb5a4bf46fa215d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233d12a18ec04f49bf8e08f72e461ff5 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a96058889d8447dca6dbcf72d2010967 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/035415f50024422b9a6a66b568536f9f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461175c6767544189e152cd0613698ff : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73ba1baf41924f41ab6e89966293f30b : SUCCESS in 3m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4f90e2e790574a2fb31df7bb6db6b5f7 : SUCCESS in 39m 17s\n- zun-tempest-py3-docker-sql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f04844444834d40a17f01dc1f2ea93d : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b3009e7e14619aea67cf4eba04a9fd2bd9b9a9bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 15:30:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5341c82572844b861c9617f9fd7301d4a56c64c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 15:31:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de1e3152a6dc456182b7825b0b0d9098\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e691b804e04beeba7f49f984899a9c : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e68db148f2e4e5298d57eafb02f06ba : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"582dd99d97335e74e99e565cbf0b89e7120bfb3b","revisions":{"582dd99d97335e74e99e565cbf0b89e7120bfb3b":{"kind":"REWORK","_number":1,"created":"2025-06-30 06:10:53.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/25/953725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/25/953725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/25/953725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/25/953725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/25/953725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/25/953725/1"}}},"commit":{"parents":[{"commit":"9cd8458c2233d8a631bf5025fdb31babddbc11b8","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9cd8458c2233d8a631bf5025fdb31babddbc11b8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2025-06-27 10:25:46.000000000","tz":0},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2025-06-30 06:10:53.000000000","tz":0},"subject":"Migrate from wsgi scripts to module paths","message":"Migrate from wsgi scripts to module paths\n\nsetuptools does not build wsgi-scripts in editable wheels anymore,\nbreaking devstack. Migrate to module paths as per governance [1].\n\n[1]https://governance.openstack.org/tc/goals/proposed/migrate-from-wsgi-scripts-to-module-paths.html\n\nChange-Id: Ibd4bc5214c55f1379738a8df0c2e3a39a6e90df0\n(cherry picked from commit fa04137529a6fb8eb3de3455c916340042a72468)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/582dd99d97335e74e99e565cbf0b89e7120bfb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/582dd99d97335e74e99e565cbf0b89e7120bfb3b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
