)]}'
{"id":"openstack%2Fzun~995022","triplet_id":"openstack%2Fzun~master~I25ea8fa362ca7622f1742fa161cf76d53ef85601","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-22 12:43:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-22 13:10:14.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"13039":{"account":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"last_update":"2026-08-22 13:31:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25ea8fa362ca7622f1742fa161cf76d53ef85601","subject":"Fix capsule type mismatch when concatenating container lists","status":"MERGED","created":"2026-06-26 09:59:13.000000000","updated":"2026-08-22 13:32:20.000000000","submitted":"2026-08-22 13:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995022","meta_rev_id":"25535f4a95ef520bb2d3c8f3dddf1004a1cbb819","_number":995022,"virtual_id_number":995022,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-22 13:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-08-22 12:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2026-06-30 15:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-26 11:41:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-08-22 12:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 10:52:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 11:41:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-26 11:41:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-06-30 15:34:42.000000000","updated_by":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"reviewer":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"state":"REVIEWER"}],"messages":[{"id":"5ee74765d67ce7aa796fc31228f1c37e704004e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-06-26 09:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0095479d4ea102e53081ad6eac7ca7541538ec4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:52:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f592836e525e474d909c1befa9dbdba9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b47468401686484395d87be768b3e394 : FAILURE in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9c5f225a2d40b7baea29734cfeadac : FAILURE in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca3d756dc98e4045a8bf01b76664d499 : FAILURE in 7m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39599634a0e242598699f3899b9ee555 : FAILURE in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/30739fdb197f42b89951bbd53946513e : FAILURE in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321f52240e244fe8ba8d27f2ee121754 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b0a93d8a2e1454082c572415001a78a : SUCCESS in 4m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a2320d4c4d3544539ec7a72853f3961d : SUCCESS in 44m 45s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/9142dd170d554848afd162c2c73de59e : SUCCESS in 45m 48s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/c967b9faa3a643c99f086cc56053dfd4 : SUCCESS in 47m 37s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/c78cff5dfc4743ec9b9b06a0de7aeb70 : FAILURE in 14m 11s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b321eda702514c83be3cd705083bff5e : SUCCESS in 41m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b692c7d45713d5a058fcc8ba0343c6600683ace8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:17:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1d648266452189fcb91946fbf0add1c1ff919eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:18:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"00d5b97644179bc3c7895a91e7076cb4430fe6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:30:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8bb648e98307c8ce9d5d9aefe0307863743c3a1c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:41:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1b6441b7d996bfe2ca85883d17338d73b87e7460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 12:21:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffbc1dfe90f48b3a9e6dab47e5faf5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c88c0f2b2344bc883f10e03675d0d4f : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ab71f758e246b79dcf9157d79d799c : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bf32752da604415848f890eb9aae3ed : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49f2c8a27dbc43429a9f83c5f1e42957 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9b26d1a06b4408fb92b169685a2d251 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ece156c29c4aaa818a96875462510c : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae2416bcfc3453288b8fb92cbc9b4c5 : SUCCESS in 9m 17s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a3af505a8d6b46d49a6a21223cee4352 : SUCCESS in 46m 16s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/86b8d05a99ea49b992f4d9a466168edc : SUCCESS in 36m 50s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d5b1cd92b1fa4717a29bfe6ab74c04d5 : SUCCESS in 46m 12s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/eb0357e3895c4ecdae00da0b60a8f27b : FAILURE in 17m 17s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/590e94d7670942a98abcb9ba77dcc0c2 : SUCCESS in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7192b504e324c28795034edaaf501fe34242566a","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-06-30 15:34:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9ceeb6045e982cbeef5e3ec4b15752c3a8f47686","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-08-22 12:43:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cddf943799bc127b0ab0bf1be424a3b49474eafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 12:44:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b6f43b393db61678bbe6bbd7081113fb27eef246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 13:31:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f03753d147f4a4199165637bd36a53b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924eba63b9f849238d56e8c64cd4c4c0 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17b1d5fd83cc40baa0431fa3eb247028 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/562e9ab7c06c4326a7bb32abb0d1b8fa : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2ef2eb81d44e8e9dba66e5fc8d7149 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1898af2bbf0d4f068a266a90175979e3 : SUCCESS in 6m 48s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0ba50c3db7454efca87726cf569b5189 : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":4},{"id":"c24852c0141705d01fe023249bab9d7bd4b77201","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 13:31:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"25535f4a95ef520bb2d3c8f3dddf1004a1cbb819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 13:32:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efcf89b284004b9f8e6d6aa259bf77fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cb3daa9119409a94638de85aa531c6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"db0fec71fa95caaa90d853f40ae4283cbb3c6bf0","revisions":{"f82c8ace6b0ae928aef1a537f2f8e765fd2f5e6b":{"kind":"REWORK","_number":1,"created":"2026-06-26 09:59:13.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/22/995022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/995022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/995022/1"}}},"commit":{"parents":[{"commit":"e79265e88ff986c608e6ef52730014421a42a21e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e79265e88ff986c608e6ef52730014421a42a21e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-26 09:57:09.000000000","tz":120},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-26 09:59:03.000000000","tz":120},"subject":"Fix capsule type mismatch when concatenating container lists","message":"Fix capsule type mismatch when concatenating container lists\n\noslo.versionedobjects introduced CoercedList which enforces element\ntype coercion on __add__. Since CapsuleContainer and\nCapsuleInitContainer are different types, concatenating their typed\nlists with + raises ValueError. Fix by converting to plain Python\nlists before concatenating.\n\nCloses-Bug: #2158372\nChange-Id: I25ea8fa362ca7622f1742fa161cf76d53ef85601\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f82c8ace6b0ae928aef1a537f2f8e765fd2f5e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f82c8ace6b0ae928aef1a537f2f8e765fd2f5e6b"}]},"branch":"refs/heads/master"},"1fe7b9dbefcf33d9d513579de237dd79fb0da306":{"kind":"REWORK","_number":2,"created":"2026-06-26 11:17:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/995022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/995022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/995022/2"}}},"commit":{"parents":[{"commit":"e79265e88ff986c608e6ef52730014421a42a21e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e79265e88ff986c608e6ef52730014421a42a21e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-26 09:57:09.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 11:17:12.000000000","tz":0},"subject":"Fix capsule type mismatch when concatenating container lists","message":"Fix capsule type mismatch when concatenating container lists\n\noslo.versionedobjects introduced CoercedList which enforces element\ntype coercion on __add__. Since CapsuleContainer and\nCapsuleInitContainer are different types, concatenating their typed\nlists with + raises ValueError. Fix by converting to plain Python\nlists before concatenating.\n\nCloses-Bug: #2158372\nChange-Id: I25ea8fa362ca7622f1742fa161cf76d53ef85601\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1fe7b9dbefcf33d9d513579de237dd79fb0da306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1fe7b9dbefcf33d9d513579de237dd79fb0da306"}]},"branch":"refs/heads/master"},"2e0c18dfbb57f9ed4618218f97399d16b596cebe":{"kind":"REWORK","_number":3,"created":"2026-06-26 11:18:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/995022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/995022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/995022/3"}}},"commit":{"parents":[{"commit":"e79265e88ff986c608e6ef52730014421a42a21e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e79265e88ff986c608e6ef52730014421a42a21e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-26 09:57:09.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 11:18:11.000000000","tz":0},"subject":"Fix capsule type mismatch when concatenating container lists","message":"Fix capsule type mismatch when concatenating container lists\n\noslo.versionedobjects introduced CoercedList which enforces element\ntype coercion on __add__. Since CapsuleContainer and\nCapsuleInitContainer are different types, concatenating their typed\nlists with + raises ValueError. Fix by converting to plain Python\nlists before concatenating.\n\nCloses-Bug: #2158372\nChange-Id: I25ea8fa362ca7622f1742fa161cf76d53ef85601\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2e0c18dfbb57f9ed4618218f97399d16b596cebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2e0c18dfbb57f9ed4618218f97399d16b596cebe"}]},"branch":"refs/heads/master"},"db0fec71fa95caaa90d853f40ae4283cbb3c6bf0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-26 11:30:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/995022/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/995022/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/995022/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/995022/4"}}},"commit":{"parents":[{"commit":"925e74f1476ee72ae5fe50fa6d958065dcca002d","subject":"Drop reference to legacy keystonemiddleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/925e74f1476ee72ae5fe50fa6d958065dcca002d"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-26 09:57:09.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 11:30:13.000000000","tz":0},"subject":"Fix capsule type mismatch when concatenating container lists","message":"Fix capsule type mismatch when concatenating container lists\n\noslo.versionedobjects introduced CoercedList which enforces element\ntype coercion on __add__. Since CapsuleContainer and\nCapsuleInitContainer are different types, concatenating their typed\nlists with + raises ValueError. Fix by converting to plain Python\nlists before concatenating.\n\nCloses-Bug: #2158372\nChange-Id: I25ea8fa362ca7622f1742fa161cf76d53ef85601\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/db0fec71fa95caaa90d853f40ae4283cbb3c6bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/db0fec71fa95caaa90d853f40ae4283cbb3c6bf0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e79265e88ff986c608e6ef52730014421a42a21e","ours":"2e0c18dfbb57f9ed4618218f97399d16b596cebe","theirs":"925e74f1476ee72ae5fe50fa6d958065dcca002d","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
