)]}'
{"id":"openstack%2Fzun~995037","triplet_id":"openstack%2Fzun~stable%2F2026.1~I2f1ce306e3eddaa4f583b25b324e8d78f3037261","project":"openstack/zun","branch":"stable/2026.1","topic":"fix_devstack_upload_images-stable/2026.1","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-29 19:47:33.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-26 12:40:53.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I2f1ce306e3eddaa4f583b25b324e8d78f3037261","subject":"devstack: Fix bash conditional in upload_images","status":"MERGED","created":"2026-06-26 11:50:16.000000000","updated":"2026-08-29 19:48:32.000000000","submitted":"2026-08-29 19:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995037-fix_devstack_upload_images-stable/2026.1","cherry_pick_of_change":839134,"cherry_pick_of_patch_set":2,"meta_rev_id":"54d88043952be528c74eb379c51e96f7ca1ee3fd","_number":995037,"virtual_id_number":995037,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-29 19:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-29 18:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-29 18:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 11:50:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"CC"},{"updated":"2026-06-26 11:50:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-06-26 12:37:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff8f6ef165d1e81bf1b8714da979baf1de15fbdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:50:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"802cb421f2c0ea9714fe10ddae681b8c9916f304","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 11:50:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6dd07e79bc30047324ca050d443ef7ab7445bfe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 12:37:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d78c89fe8a614885b1bb472b15a70fe9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ad3564f8324b5593ef5397c71d1856 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008adec19939466aa948061755666963 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/793cc3c5567849cf8011dfa5e5ea8aae : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcf9bcdc3d2d436cbd8079a3702ba7c2 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b360852d0d944b78d3e5a8c6f798199 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6ffabf977e44899ae0fa946a722526 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/134b981c0bc34ae1bf9e842ef14e29d4 : SUCCESS in 4m 25s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/df5bf943ee3e4f23a9348c3408abad79 : FAILURE in 29m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/d3b76f7b3fd640f18959c0b1c90255e9 : FAILURE in 29m 16s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/258f0042e43b4ba8a71294f13de3169f : FAILURE in 17m 30s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/82ca1dc942ec4421af4fd1c32c097fbd : FAILURE in 27m 19s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/81a9e2c3f4af43219f1d46aea138c71c : SUCCESS in 41m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b4a7ad93c8131f9a11403b4fefc9166dcf918a2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 12:40:53.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"21c8f6dbea6d37675afdf9e97609504124c08245","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 12:47:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"706c9305c6138e73214b4b86c4c3d8597cd3d5bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:21:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ab94acecfba7441083f456a7f5179d61159eb338","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:21:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"62cf3b4db2d9a2c35310af72f81f18bb047ca678","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:21:42.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"210fb1b918e66ff9d2d1fec2192b1d27d7b61212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:21:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"be854d837bd9d16b21cedd2927d8673ea01d80f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 21:17:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d710052d0b294b71ad5f0da022693d9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8091e665feb546558a7dc71a6dc9b94a : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a2bbb6f56840a18e62b7e03aa915e0 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1ca490e8ec241a089215022375526e7 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d503b913ab1f4bb0852277de7899cea8 : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e01eadcef0804888bd4e116374633778 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e08f0eec85947feb7bd360d48829f0f : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec36a00152f420f97f6349b3a93c721 : SUCCESS in 5m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c8fbb1d3ea1e4f7191fb792a48cb64ea : SUCCESS in 23m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/e274ccbf62624c82b7a8e81ff156788a : SUCCESS in 48m 34s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/822be6bb8c8747e6889978dc622affb4 : SUCCESS in 24m 45s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/5d402634b08548e89c5ad715a1a86026 : FAILURE in 16m 24s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/d12f2d7468794c99b4c75ec320fac6e4 : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7a35b37f62e1aceb2f34625918be7246cadc0a04","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 18:57:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"973e1cce535cafc82401340ba6d47eb213836a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 18:57:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b074437c9cb03f9401716cad8e1c3cf58ffb8cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 19:47:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29949cce3292460c9b98fbb1b942ed66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a5219e8c8c4cfeae6864f2f4cfa771 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232c666a250347e19c6bcdcbe4f7ba8b : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f425c56dd64841af962f72fc3d175b : SUCCESS in 3m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/482fffb1926045a1981bb5be60c71941 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6b0edc54cfe955790ce091428291320b239a0460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 19:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"54d88043952be528c74eb379c51e96f7ca1ee3fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 19:48:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807014aee98f461cb8e5b81f74ad8da5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ea370541b34b18b550e99e51cc2aa8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"936152fa167d2cbeffb77ba71a66849f952da155","revisions":{"c3672b583fee807f1385e9714cd4d040172634da":{"kind":"REWORK","_number":1,"created":"2026-06-26 11:50:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/995037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/995037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/995037/1"}}},"commit":{"parents":[{"commit":"11475f237dc36d72c7e1b02040c7a4edac9ce2ae","subject":"sqlalchemy: Use built-in declarative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/11475f237dc36d72c7e1b02040c7a4edac9ce2ae"}]}],"author":{"name":"Martin André","email":"martin.andre@gmail.com","date":"2022-04-24 15:43:06.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 11:50:16.000000000","tz":0},"subject":"Devstack: Fix bash conditional in upload_images","message":"Devstack: Fix bash conditional in upload_images\n\nWe need to evaluate the expression otherwise bash tries to execute it\nand it causes the following error in the logs:\n\n   ++ /opt/stack/zun/devstack/lib/zun:upload_images:139 :   docker \u003d\u003d docker\n   docker: \u0027\u003d\u003d\u0027 is not a docker command.\n   See \u0027docker --help\u0027\n\nChange-Id: I2f1ce306e3eddaa4f583b25b324e8d78f3037261\nSigned-off-by: Martin André \u003cmartin.andre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c3672b583fee807f1385e9714cd4d040172634da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c3672b583fee807f1385e9714cd4d040172634da"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"2e8825a05f397b01b760670d606f3c3cb2a8d3cc","ours":"11475f237dc36d72c7e1b02040c7a4edac9ce2ae","theirs":"421aad3328659c53f0507fd8effd908848f5b801","merge_strategy":"recursive","contains_conflicts":false}},"94f4acbde512187f38717a2e60d8c90c6f540d1a":{"kind":"REWORK","_number":2,"created":"2026-06-26 20:21:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/995037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/995037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/995037/2"}}},"commit":{"parents":[{"commit":"11475f237dc36d72c7e1b02040c7a4edac9ce2ae","subject":"sqlalchemy: Use built-in declarative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/11475f237dc36d72c7e1b02040c7a4edac9ce2ae"}]}],"author":{"name":"Martin André","email":"martin.andre@gmail.com","date":"2022-04-24 15:43:06.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 20:21:31.000000000","tz":0},"subject":"Devstack: Fix bash conditional in upload_images","message":"Devstack: Fix bash conditional in upload_images\n\nWe need to evaluate the expression otherwise bash tries to execute it\nand it causes the following error in the logs:\n\n   ++ /opt/stack/zun/devstack/lib/zun:upload_images:139 :   docker \u003d\u003d docker\n   docker: \u0027\u003d\u003d\u0027 is not a docker command.\n   See \u0027docker --help\u0027\n\nChange-Id: I2f1ce306e3eddaa4f583b25b324e8d78f3037261\nSigned-off-by: Martin André \u003cmartin.andre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/94f4acbde512187f38717a2e60d8c90c6f540d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/94f4acbde512187f38717a2e60d8c90c6f540d1a"}]},"branch":"refs/heads/stable/2026.1"},"488b10989cc57ad21c63f95fcd1b283a2465330f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-26 20:21:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/995037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/995037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/995037/3"}}},"commit":{"parents":[{"commit":"11475f237dc36d72c7e1b02040c7a4edac9ce2ae","subject":"sqlalchemy: Use built-in declarative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/11475f237dc36d72c7e1b02040c7a4edac9ce2ae"}]}],"author":{"name":"Martin André","email":"martin.andre@gmail.com","date":"2022-04-24 15:43:06.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 20:21:38.000000000","tz":0},"subject":"devstack: Fix bash conditional in upload_images","message":"devstack: Fix bash conditional in upload_images\n\nWe need to evaluate the expression otherwise bash tries to execute it\nand it causes the following error in the logs:\n\n   ++ /opt/stack/zun/devstack/lib/zun:upload_images:139 :   docker \u003d\u003d docker\n   docker: \u0027\u003d\u003d\u0027 is not a docker command.\n   See \u0027docker --help\u0027\n\nChange-Id: I2f1ce306e3eddaa4f583b25b324e8d78f3037261\nSigned-off-by: Martin André \u003cmartin.andre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/488b10989cc57ad21c63f95fcd1b283a2465330f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/488b10989cc57ad21c63f95fcd1b283a2465330f"}]},"branch":"refs/heads/stable/2026.1","description":"Edit commit message"},"936152fa167d2cbeffb77ba71a66849f952da155":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-26 20:21:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/995037/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/37/995037/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/37/995037/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/37/995037/4"}}},"commit":{"parents":[{"commit":"8e59c2fec82edc4c3d75dd828a0e63b3bae39e71","subject":"devstack: use new image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8e59c2fec82edc4c3d75dd828a0e63b3bae39e71"}]}],"author":{"name":"Martin André","email":"martin.andre@gmail.com","date":"2022-04-24 15:43:06.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 20:21:58.000000000","tz":0},"subject":"devstack: Fix bash conditional in upload_images","message":"devstack: Fix bash conditional in upload_images\n\nWe need to evaluate the expression otherwise bash tries to execute it\nand it causes the following error in the logs:\n\n   ++ /opt/stack/zun/devstack/lib/zun:upload_images:139 :   docker \u003d\u003d docker\n   docker: \u0027\u003d\u003d\u0027 is not a docker command.\n   See \u0027docker --help\u0027\n\nChange-Id: I2f1ce306e3eddaa4f583b25b324e8d78f3037261\nSigned-off-by: Martin André \u003cmartin.andre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/936152fa167d2cbeffb77ba71a66849f952da155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/936152fa167d2cbeffb77ba71a66849f952da155"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"11475f237dc36d72c7e1b02040c7a4edac9ce2ae","ours":"488b10989cc57ad21c63f95fcd1b283a2465330f","theirs":"8e59c2fec82edc4c3d75dd828a0e63b3bae39e71","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
