)]}'
{"id":"openstack%2Fzun~995142","triplet_id":"openstack%2Fzun~master~Ieda725b6bc0e096eeee23b18d985d52ba7516ac8","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-16 03:02:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-17 11:09:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ieda725b6bc0e096eeee23b18d985d52ba7516ac8","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2026-06-26 19:56:12.000000000","updated":"2026-08-17 13:11:31.000000000","submitted":"2026-08-17 13:10:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995142","meta_rev_id":"0d872d5a35651b7a7f0395f73d4cc5709a742b43","_number":995142,"virtual_id_number":995142,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-17 13:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-08-16 03:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-08-16 03:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 20:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 22:03:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"ffbe3e5966d8f1316cef21f61b0419ebe83c2125","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 19:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d51f3aa0c7387407f20cdce40105b6c98785a74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 19:59:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69308bdf90d0c571c015d1ffa77f10749365c323","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:10:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bcd8227f716c1c4310783dad34e82317b9f25cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 20:49:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dc36db40526484d819e2472b3940d72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0872a9b38ab44cc9bf77ba41c7aab0cf : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc842815391469d8b3978c793cc60ef : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9dccb6819ed4934ba4f1bdccce185bd : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4abe3e3918d4eeaabeac0a943a0c47c : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae7d6e5e6fb04d828e52e8ee08e5b24a : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b328899e9f1042b5accc5eea6d846b7a : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b240a6eeb94b33abdc3c42786b05cf : SUCCESS in 5m 59s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/13fb796e0d644787bdc0fc33cbc0fe67 : FAILURE in 27m 08s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4c28867934514adc8696f9b756efc0e6 : FAILURE in 28m 36s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/9c97dda8363b40d1a4aa299a5ddf45e4 : FAILURE in 31m 39s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d845bfe42546408f91728c140cfa7ae0 : FAILURE in 29m 50s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2c424e26585c4ad4865969a0f4132e60 : FAILURE in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e00ebafe736ab0992de7f6df600941cd606a35b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 20:57:02.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"87a5835bf5ca0bce2583068fbeb569f3ccc9e142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:02:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4ff2d1a3434b46859d94cdee8de217\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ec0ee75cbfb495c870e5f966ab5aa8c : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bdddced9a93648e1afceffdb78958663 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac88ef569ad341d39397679b1330c6a8 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5e8b52cc4946e1842d663b4d742520 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/329fe212436f4dedb83c67867c6e9f6a : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7baaa66968bf4047903a887a93445928 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d48e9eac28643289960ef31fe5553a8 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd6808fa9f54676a59c93df97c1bb64 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e79fe30f25c14d4291152987296dd395 : SUCCESS in 6m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9cb4c382f3094b9692f97e3492b89210 : SUCCESS in 44m 49s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/71572ca4e49f48cfa77a5195c82a1ca9 : SUCCESS in 45m 40s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/98beb7b3fc934e99bb02e470f18b077e : SUCCESS in 52m 14s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d8da21372ebd45389940554654097c9b : FAILURE in 19m 34s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/24ae40214de3485f8b9635aeb4987f1d : SUCCESS in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd110eb4d72e08c30ec5625f804e6be37a76d6fe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 22:03:25.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"caa7c5a01fe0534afabdb0c6e40d4885a55101ca","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-08-16 03:02:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b9acd84c8d76cd0648a0b60f44e4148d2476e590","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 08:01:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0bac416e49be29dbd73b49cde03592d581cf5eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 08:53:53.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9beed4d4fda64596b23ef5ce1c2a9134\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7608b47f40394acda9dfbb245677f62d : SUCCESS in 3m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/889e180b6749467d8004cfa652218088 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28d459ebbbd4a6aac504ff12d528ffb : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fac8f38cfe3d4ddf881cf6d79db03c7d : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afd352ca63714621a1412187bd74d207 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dedf3a8dae644ed29fdec41311a5ef44 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba356b0aa064e269c8907381d9676ea : SUCCESS in 5m 53s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0c45af9650ac49c5a25075f81ff1f81d : FAILURE in 28m 37s","accounts_in_message":[],"_revision_number":4},{"id":"baf25186aa9822261abb4fb4bcc653e69f52344f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-17 11:09:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dcfff38180fe129b86f5fba8d117ce5586579396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 12:09:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae93fa7ffb34d2998f4c23fb3b052e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/277c168ad0264dc2aca30c27fce3027d : SUCCESS in 3m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3ffbf5cadf34bc3976aa9550f3d7086 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/898d9e29b58d418aa1358f7dec660d86 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f59aed9a784fe7bcffd524d7d529ad : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/867e255ab758480eb07f23116f15c633 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a17f9bed81474c3e9dd4d036e8514d0e : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1415ea34adb24204be04159429baef7a : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b515f8b25034704bcd966456d588655 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7fcb509878f446cacef9c5f8ae4f528 : SUCCESS in 6m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5c5c5568b0f9433f9b970e0cfc2f4638 : SUCCESS in 45m 35s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/4df8583916144cf3a69f5b8f2bf090d9 : SUCCESS in 36m 19s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b57226a7719b4b729a4f592af2a10ca5 : SUCCESS in 47m 20s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/129038dab3634eed8fb30e1588af02fc : FAILURE in 13m 00s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/b0e12a3162f04033bf2edca23b34160b : SUCCESS in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"634846b939633c6a387314568836e8009538d111","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 12:09:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a98c1b15195c31cb733d2ff642fceecdbc82c016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 13:10:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616b2f7d13084d1ea88f33eb959b5d9e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3dad85ef5b5f4d9db69c8008ddb6fa06 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1755747d79bf411bba25ac8e773ceed9 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6027a60f2b473fab1212c6e4511f1a : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/810834c9ba234837bcb54a0d1db9900c : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ab09d6aefc946e3aec140a25b88de74 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ccff9666e645a2b4e303b9adecaa24 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8419f034f72d48638e5df672b6c45e73 : SUCCESS in 4m 01s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9f3766ddb9f045e1b59e879a99a971e6 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":4},{"id":"dded9f7b39b24476e8fa50a6ee6f7751e0cb8973","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 13:10:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0d872d5a35651b7a7f0395f73d4cc5709a742b43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 13:11:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93157173db7843669a0935ad41abeb6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a6299a3e854859abfea7f862c08c4a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"df69a9c2837a589fc19110342c0d95dc41df99e9","revisions":{"be1553de557696099c8e9c6c1f15f98857a93eac":{"kind":"REWORK","_number":1,"created":"2026-06-26 19:56:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/995142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/995142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/995142/1"}}},"commit":{"parents":[{"commit":"e79265e88ff986c608e6ef52730014421a42a21e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e79265e88ff986c608e6ef52730014421a42a21e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:56:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:56:12.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ieda725b6bc0e096eeee23b18d985d52ba7516ac8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/be1553de557696099c8e9c6c1f15f98857a93eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/be1553de557696099c8e9c6c1f15f98857a93eac"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"3b05419b3446b0877a5d9d54764ebff76ae22c27":{"kind":"REWORK","_number":2,"created":"2026-06-26 19:59:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/995142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/995142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/995142/2"}}},"commit":{"parents":[{"commit":"e79265e88ff986c608e6ef52730014421a42a21e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/e79265e88ff986c608e6ef52730014421a42a21e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:56:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:59:48.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ieda725b6bc0e096eeee23b18d985d52ba7516ac8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3b05419b3446b0877a5d9d54764ebff76ae22c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3b05419b3446b0877a5d9d54764ebff76ae22c27"}]},"branch":"refs/heads/master"},"83103705ad4cd38e388c09777522bc2dbd89edce":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-26 20:10:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/995142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/995142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/995142/3"}}},"commit":{"parents":[{"commit":"925e74f1476ee72ae5fe50fa6d958065dcca002d","subject":"Drop reference to legacy keystonemiddleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/925e74f1476ee72ae5fe50fa6d958065dcca002d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:56:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 20:10:42.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ieda725b6bc0e096eeee23b18d985d52ba7516ac8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/83103705ad4cd38e388c09777522bc2dbd89edce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/83103705ad4cd38e388c09777522bc2dbd89edce"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e79265e88ff986c608e6ef52730014421a42a21e","ours":"3b05419b3446b0877a5d9d54764ebff76ae22c27","theirs":"925e74f1476ee72ae5fe50fa6d958065dcca002d","merge_strategy":"recursive","contains_conflicts":false}},"df69a9c2837a589fc19110342c0d95dc41df99e9":{"kind":"REWORK","_number":4,"created":"2026-07-07 20:57:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/995142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/995142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/995142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/995142/4"}}},"commit":{"parents":[{"commit":"925e74f1476ee72ae5fe50fa6d958065dcca002d","subject":"Drop reference to legacy keystonemiddleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/925e74f1476ee72ae5fe50fa6d958065dcca002d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:56:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-07 20:57:02.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nChange-Id: Ieda725b6bc0e096eeee23b18d985d52ba7516ac8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df69a9c2837a589fc19110342c0d95dc41df99e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/df69a9c2837a589fc19110342c0d95dc41df99e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
