)]}'
{".zuul.yaml":[{"author":{"_account_id":15435,"name":"Al Bailey","email":"albailey1974@gmail.com","username":"albailey"},"change_message_id":"e5d78510aa3142a3cbb54c00a454ab927a7562cb","unresolved":false,"context_lines":[{"line_number":25,"context_line":""},{"line_number":26,"context_line":"# Perform just a build"},{"line_number":27,"context_line":"- job:"},{"line_number":28,"context_line":"    name: stx-fault-build"},{"line_number":29,"context_line":"    parent: tox"},{"line_number":30,"context_line":"    description: Run build for fault"},{"line_number":31,"context_line":"    irrelevant-files:"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"3f79a3b5_943fc782","line":28,"updated":"2018-12-10 14:19:28.000000000","message":"Once you get the build script working as expected, would we want to add this to the gate as well?","commit_id":"930641793ba1c106d3c7d30d42c44b3604df1f92"}]}
