)]}'
{"id":"x%2Fdevstack-plugin-tobiko~914884","triplet_id":"x%2Fdevstack-plugin-tobiko~master~I8b6785cda7efc60a86a9365ef29b0ec0d6515025","project":"x/devstack-plugin-tobiko","branch":"master","topic":"OSPRH-2772","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-04-02 17:39:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b6785cda7efc60a86a9365ef29b0ec0d6515025","subject":"Skip Stateless SG tests also in the stable/2023.1 branch","status":"MERGED","created":"2024-04-02 13:01:37.000000000","updated":"2024-04-02 17:39:04.000000000","submitted":"2024-04-02 17:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914884-OSPRH-2772","meta_rev_id":"58de933352d21a883438805d24eb5aa4329e7edb","_number":914884,"virtual_id_number":914884,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 17:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"all":[{"value":2,"date":"2024-04-02 14:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"all":[{"value":1,"date":"2024-04-02 14:53:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 14:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 14:53:30.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"}],"messages":[{"id":"389dab7956bf71010447fd0cabc71cc665194623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-02 13:01:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcbae6b770ffea3c2c33309dc69a9f5f10bfc1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdc2a07d4e04a2ea56b046f179b8a99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e4a9ff846a4613a2ad8fda7eabf5ec : SUCCESS in 3m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36975c1495d44efe88b9efbd3e126c4b : SUCCESS in 2m 20s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/ef018140a7b54e2f87a5562af443174f : SUCCESS in 41m 25s\n- devstack-tobiko-multinode https://zuul.opendev.org/t/openstack/build/6abb1e703be14e608ff3c24ac784d448 : SUCCESS in 1h 35m 14s (non-voting)\n- devstack-tobiko-centos https://zuul.opendev.org/t/openstack/build/60b9de4ca9904bd2ae8e6248d4ab26a6 : RETRY_LIMIT in 9m 05s (non-voting)\n- devstack-tobiko-designate https://zuul.opendev.org/t/openstack/build/3ba390d3c40a42b0b93b0904fef29191 : SUCCESS in 43m 15s\n- devstack-tobiko-ovs-designate https://zuul.opendev.org/t/openstack/build/5a936ab1a3c645fc98eb31a3f262a26c : SUCCESS in 54m 29s\n- devstack-tobiko-functional https://zuul.opendev.org/t/openstack/build/f0edf02a3a0248a6bc62cef6b3ccca44 : FAILURE in 53m 29s (non-voting)\n- devstack-tobiko-heat https://zuul.opendev.org/t/openstack/build/adb9b03a60f848d989b751aeb483b27f : SUCCESS in 51m 56s\n- devstack-tobiko-minimal https://zuul.opendev.org/t/openstack/build/499f68a2e89a4dbc85cb6abc7522358b : SUCCESS in 49m 50s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/392ae8960e564409af476f4a91393975 : SUCCESS in 39m 09s\n- devstack-tobiko-ovs-neutron https://zuul.opendev.org/t/openstack/build/338dfabbe15e46a3a12721de9942b941 : SUCCESS in 46m 23s\n- devstack-tobiko-nova https://zuul.opendev.org/t/openstack/build/ad84f9f2f1eb4dc4a87217fa4cd5f5a0 : SUCCESS in 27m 29s\n- devstack-tobiko-multinode-nova https://zuul.opendev.org/t/openstack/build/0d680a555e9144a4b8c7dba9bad28d9a : SUCCESS in 1h 17m 30s\n- devstack-tobiko-octavia https://zuul.opendev.org/t/openstack/build/7bba7a9686284f07bd1b4cf9504e6ff2 : SUCCESS in 39m 35s\n- devstack-tobiko-ovs-octavia https://zuul.opendev.org/t/openstack/build/cf22eebdeaf84d9eb527a52c1d3cb797 : SUCCESS in 48m 30s (non-voting)\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/eafe1fee024b4161b7018b85e58e1272 : SUCCESS in 51m 51s\n- devstack-tobiko-sanity https://zuul.opendev.org/t/openstack/build/f42e443064a341f19b48f54e6f84df77 : SUCCESS in 23m 32s\n- devstack-tobiko-storage https://zuul.opendev.org/t/openstack/build/9f62e5cf9b7c4f4d85cb97cb1325409b : SUCCESS in 34m 45s\n- devstack-tobiko-storage-ceph https://zuul.opendev.org/t/openstack/build/85d58c0133ac4cfbb0ee3a115a694f3f : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b92b38300d6062107816f2c94a6a948804d44155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-02 14:51:45.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"440c7426b13d36e5a7d8b301a0b3e7d7e8d07d5f","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2024-04-02 14:53:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b50a5fb92239d990d33657df3540cd32ff0b05a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:35:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9da31645bb46d394cf08726f65837f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda71b117d1d423d8a0174d699e9fe49 : SUCCESS in 4m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/126166051cdb4dfda61ee12d95001f07 : SUCCESS in 3m 03s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/755ad73be4aa4f65aa2893a675ce73b9 : SUCCESS in 1h 00m 43s\n- devstack-tobiko-multinode https://zuul.opendev.org/t/openstack/build/f169b842bc6f49ccbf0ef5a82e44fcf2 : SUCCESS in 1h 34m 21s (non-voting)\n- devstack-tobiko-centos https://zuul.opendev.org/t/openstack/build/b7e14e96dab34fa3b58790ac1c0fabcf : RETRY_LIMIT in 7m 17s (non-voting)\n- devstack-tobiko-designate https://zuul.opendev.org/t/openstack/build/873d19b9958140fabeeea7b47536661e : SUCCESS in 52m 45s\n- devstack-tobiko-ovs-designate https://zuul.opendev.org/t/openstack/build/fdf0e0317dcc4a2f9bb70d8c0275fde3 : SUCCESS in 56m 23s\n- devstack-tobiko-functional https://zuul.opendev.org/t/openstack/build/c5f765ce6cb9470b88531123088e53d8 : SUCCESS in 41m 30s (non-voting)\n- devstack-tobiko-heat https://zuul.opendev.org/t/openstack/build/3195b928e8fd47c4914f410f3f89a06c : SUCCESS in 48m 22s\n- devstack-tobiko-minimal https://zuul.opendev.org/t/openstack/build/dfedea8d17964c24acbed9e88da0f79c : SUCCESS in 1h 07m 30s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/8d09218eeaf4477eab71d37d535b2b6d : SUCCESS in 48m 59s\n- devstack-tobiko-ovs-neutron https://zuul.opendev.org/t/openstack/build/560c4b846b394972bdb25d4fddb3e990 : SUCCESS in 47m 44s\n- devstack-tobiko-nova https://zuul.opendev.org/t/openstack/build/092b22e78ac94d2199c4cced52d3cc29 : SUCCESS in 28m 49s\n- devstack-tobiko-multinode-nova https://zuul.opendev.org/t/openstack/build/7a7fa8b96de7484d851ba53b74ba1225 : SUCCESS in 1h 12m 51s\n- devstack-tobiko-octavia https://zuul.opendev.org/t/openstack/build/930de42e746f4b5d9eb0088acfa8a6cc : SUCCESS in 57m 42s\n- devstack-tobiko-ovs-octavia https://zuul.opendev.org/t/openstack/build/d3dfc89fb45f43319a282e30c6a1a4ad : SUCCESS in 53m 21s (non-voting)\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/29362d9c0d10486f89f75fc7de91f347 : SUCCESS in 1h 01m 10s\n- devstack-tobiko-sanity https://zuul.opendev.org/t/openstack/build/2a287d4f0e304254a9fcbf51c17631cf : SUCCESS in 21m 43s\n- devstack-tobiko-storage https://zuul.opendev.org/t/openstack/build/20f5597b248c47c598e365c8f5ae2394 : SUCCESS in 51m 42s\n- devstack-tobiko-storage-ceph https://zuul.opendev.org/t/openstack/build/1741947e11df437fb820dcfaf400f93b : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"01140d3f3bcb1d62bcd602280c1e838d1b23f8fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:35:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d439396705d8677808da8c9a3f27d3f459c24301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:39:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b3aefc81dc4933b8af55e192cc77a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf4775930e74638a9fbc6c93abfc285 : SUCCESS in 2m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69e01b09f7954340b72c35520c3d452c : SUCCESS in 2m 17s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/610de7914f6f46a2a48baf6c64baedf7 : SUCCESS in 48m 07s\n- devstack-tobiko-designate https://zuul.opendev.org/t/openstack/build/f86ba9511277442dae93d64ec7682c14 : SUCCESS in 50m 35s\n- devstack-tobiko-functional https://zuul.opendev.org/t/openstack/build/e98099b3d93f4abea87283369238ac73 : FAILURE in 37m 30s (non-voting)\n- devstack-tobiko-minimal https://zuul.opendev.org/t/openstack/build/2fd0870b04b54ef282b0eb22bd6e08ba : SUCCESS in 59m 57s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/3ae37496e7ce4dda85742aa106f57a4d : SUCCESS in 48m 09s\n- devstack-tobiko-ovs-neutron https://zuul.opendev.org/t/openstack/build/f56b656407ee4649aef935c0e1b11658 : SUCCESS in 56m 05s\n- devstack-tobiko-nova https://zuul.opendev.org/t/openstack/build/b6faf8c2ceca4a0db423460ebb528a2c : SUCCESS in 33m 05s\n- devstack-tobiko-octavia https://zuul.opendev.org/t/openstack/build/0c44198a56f84faea464fc463fecc2b8 : SUCCESS in 50m 53s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/e592dbc2e54549b0a41775591656774c : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":2},{"id":"58de933352d21a883438805d24eb5aa4329e7edb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:39:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b61a5bb6bc38c25bdcccdcc76429544e6e3d3d1","revisions":{"bfb14d76a528a95ecd9b4ede907109c9e7a53083":{"kind":"REWORK","_number":1,"created":"2024-04-02 13:01:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/914884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/devstack-plugin-tobiko","ref":"refs/changes/84/914884/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/1"}}},"commit":{"parents":[{"commit":"b357f56600f2690e93c7fc78630ee482ec6d41f0","subject":"Set correct min version of tox to be installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/b357f56600f2690e93c7fc78630ee482ec6d41f0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-02 12:58:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-02 12:58:21.000000000","tz":120},"subject":"Skip Stateless SG tests also in the stable/2023.1 branch","message":"Skip Stateless SG tests also in the stable/2023.1 branch\n\nPreviously with patch [1] we enabled stateless sg tests in tobiko as\nthis works fine in jobs which are using Ubuntu 22.04 or newer. But\nNeutron\u0027s stable/2023.1 job still runs on Ubuntu 20.04 so those tests\nshould be skipped in that branch.\n\n[1] https://review.opendev.org/c/x/devstack-plugin-tobiko/+/912588\n\nChange-Id: I8b6785cda7efc60a86a9365ef29b0ec0d6515025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/bfb14d76a528a95ecd9b4ede907109c9e7a53083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/bfb14d76a528a95ecd9b4ede907109c9e7a53083"}]},"branch":"refs/heads/master"},"7b61a5bb6bc38c25bdcccdcc76429544e6e3d3d1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-02 14:51:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/914884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/devstack-plugin-tobiko","ref":"refs/changes/84/914884/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/devstack-plugin-tobiko refs/changes/84/914884/2"}}},"commit":{"parents":[{"commit":"b357f56600f2690e93c7fc78630ee482ec6d41f0","subject":"Set correct min version of tox to be installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/b357f56600f2690e93c7fc78630ee482ec6d41f0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-02 12:58:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-02 14:51:45.000000000","tz":0},"subject":"Skip Stateless SG tests also in the stable/2023.1 branch","message":"Skip Stateless SG tests also in the stable/2023.1 branch\n\nPreviously with patch [1] we enabled stateless sg tests in tobiko as\nthis works fine in jobs which are using Ubuntu 22.04 or newer. But\nNeutron\u0027s stable/2023.1 job still runs on Ubuntu 20.04 so those tests\nshould be skipped in that branch.\n\n[1] https://review.opendev.org/c/x/devstack-plugin-tobiko/+/912588\n\nClose-bug: #2060029\n\nChange-Id: I8b6785cda7efc60a86a9365ef29b0ec0d6515025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/7b61a5bb6bc38c25bdcccdcc76429544e6e3d3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/devstack-plugin-tobiko/commit/7b61a5bb6bc38c25bdcccdcc76429544e6e3d3d1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
