)]}'
{"id":"x%2Fmicrostack~765001","triplet_id":"x%2Fmicrostack~master~I72e79a6abce089d0da8e9bb1d27f120fbd5cc49f","project":"x/microstack","branch":"master","hashtags":[],"change_id":"I72e79a6abce089d0da8e9bb1d27f120fbd5cc49f","subject":"Add command to README for retrieving admin password","status":"MERGED","created":"2020-12-01 20:10:48.000000000","updated":"2021-01-06 23:21:12.000000000","submitted":"2021-01-06 23:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765001","meta_rev_id":"9ac559acd38650696e90ee155adfd8335ec57055","_number":765001,"virtual_id_number":765001,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-06 23:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":2,"date":"2020-12-17 02:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":1,"date":"2020-12-17 02:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 21:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 23:18:49.000000000","updated_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"}],"messages":[{"id":"e31dc15aec86971d66b6b15dbc047575f6a52ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-12-01 20:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4237aeb45e20d48123d60043c476d863a0772cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 21:42:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/1654b39967bc4d32bccab6551f5ae613 : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2609649095bcdcd3af8f322d13a93a15a203b8e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-12-03 16:44:32.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"95fff5360750d48479e77a22cb309caefad96ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:08:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/40782416b1454b2bb18cc0d437569ed7 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":2},{"id":"90217896135b31a4c76664893f4c65fd8a5f0887","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2020-12-04 23:18:49.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"5572f5f5e97d3a285853face597a8afc49290c24","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2020-12-17 02:16:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b3797836154a7cd2caf09ff27d1364cbe25312dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 02:17:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8103036705c512d3cb47f01fa69f767fb9ccf7a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 05:29:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/7559d7376bf043a79693d9d588316504 : RETRY_LIMIT in 59m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4598f90f9811853fc772244c79fe7a9c52d7210a","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2020-12-17 22:08:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"920baf9437669673f7b021e0cb607ef11b7f48a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 23:14:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/426d09d2e1ed4ec0852ef59ba140f3f9 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":2},{"id":"03f2ad77a7db30d06f4b5d639a8f02c5a410268f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 23:14:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db5071d5f4bdbe3f19abcc094fe9e1ded2be6fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:54:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/0534614e0b2e430a93e13968b0ff1ee9 : FAILURE in 1h 36m 29s","accounts_in_message":[],"_revision_number":2},{"id":"bafe99a580c467e11b0a6877e8e69935ded3f06e","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2021-01-06 18:23:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9cb3bd7f544b748c32df7b6a9a9b1c5a97d902f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 20:50:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/fdf2147afa1e4b66be775d94200dcf17 : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e2c06a41b77243f4a3538f21a3314a8f9ac9ace0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 20:50:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbf93f7458d568c1060a04e3db32378a8a597a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 23:21:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- microstack-tox-snap-with-sudo https://zuul.opendev.org/t/openstack/build/b69747b7fca6454ba619e7bdeaa5f31a : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9ac559acd38650696e90ee155adfd8335ec57055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 23:21:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbe820996904248c2bff7cb678446202a20bb8e7","revisions":{"9d0def2b34282e625ff1322b9842ecf681d6f044":{"kind":"REWORK","_number":1,"created":"2020-12-01 20:10:48.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/01/765001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/microstack","ref":"refs/changes/01/765001/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/microstack refs/changes/01/765001/1"}}},"commit":{"parents":[{"commit":"a2cc37e27839e307d7b0c4d7376f0ebb0c34e60e","subject":"Fix service enablement during init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/a2cc37e27839e307d7b0c4d7376f0ebb0c34e60e"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-12-01 19:55:52.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-12-01 20:10:47.000000000","tz":-300},"subject":"Add command to README for retrieving admin password","message":"Add command to README for retrieving admin password\n\nThe admin password is no longer hardcoded so update the README\nwith details for how to retrieve the password.\n\nChange-Id: I72e79a6abce089d0da8e9bb1d27f120fbd5cc49f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/9d0def2b34282e625ff1322b9842ecf681d6f044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/9d0def2b34282e625ff1322b9842ecf681d6f044"}]},"branch":"refs/heads/master"},"cbe820996904248c2bff7cb678446202a20bb8e7":{"kind":"NO_CHANGE","_number":2,"created":"2020-12-03 16:44:32.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/01/765001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/microstack","ref":"refs/changes/01/765001/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/microstack refs/changes/01/765001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/microstack refs/changes/01/765001/2"}}},"commit":{"parents":[{"commit":"a2cc37e27839e307d7b0c4d7376f0ebb0c34e60e","subject":"Fix service enablement during init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/a2cc37e27839e307d7b0c4d7376f0ebb0c34e60e"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-12-01 19:55:52.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-12-03 16:44:31.000000000","tz":0},"subject":"Add command to README for retrieving admin password","message":"Add command to README for retrieving admin password\n\nThe admin password is no longer hardcoded so update the README\nwith details for how to retrieve the password.\n\nChange-Id: I72e79a6abce089d0da8e9bb1d27f120fbd5cc49f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/cbe820996904248c2bff7cb678446202a20bb8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/microstack/commit/cbe820996904248c2bff7cb678446202a20bb8e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
