)]}'
{"id":"x%2Fpackstack~390937","triplet_id":"x%2Fpackstack~master~Icd931f36fc14fefbcd43a266f0cd538074723adf","project":"x/packstack","branch":"master","topic":"puppet4","hashtags":[],"change_id":"Icd931f36fc14fefbcd43a266f0cd538074723adf","subject":"Support puppetlabs hiera.yaml update","status":"MERGED","created":"2016-10-26 17:14:35.000000000","updated":"2016-10-28 11:26:08.000000000","submitted":"2016-10-28 11:26:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fc6144e65b300f010c1e83c75b92ee6e919c04f","_number":390937,"virtual_id_number":390937,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2016-10-28 11:26:08.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-10-28 10:34:15.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},{"value":2,"date":"2016-10-28 10:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},{"value":1,"date":"2016-10-28 10:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-27 13:18:21.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2016-10-27 13:18:23.000000000","updated_by":{"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},"reviewer":{"_account_id":7462,"name":"Ivan Chavero","email":"ichavero@redhat.com","username":"ichavero"},"state":"REVIEWER"},{"updated":"2016-10-28 07:29:58.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2016-10-28 10:36:58.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2016-10-28 11:26:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ee43770baf7aa4ee40c896c091a4650004e66d53","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-10-26 17:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eadcfdd469e664cb879e319368e7987972fce256","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-26 17:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-packstack-integration-scenario001-tempest-centos-7 http://logs.openstack.org/37/390937/1/check/gate-packstack-integration-scenario001-tempest-centos-7/1a38ce9/ : SUCCESS in 19m 30s\n- gate-packstack-integration-scenario002-tempest-centos-7 http://logs.openstack.org/37/390937/1/check/gate-packstack-integration-scenario002-tempest-centos-7/84d0f1f/ : SUCCESS in 17m 24s\n- gate-packstack-integration-scenario003-tempest-centos-7 http://logs.openstack.org/37/390937/1/check/gate-packstack-integration-scenario003-tempest-centos-7/b2a40d4/ : SUCCESS in 21m 33s\n- gate-packstack-pep8-ubuntu-xenial http://logs.openstack.org/37/390937/1/check/gate-packstack-pep8-ubuntu-xenial/b0a5994/ : SUCCESS in 50s\n- gate-packstack-python27-ubuntu-xenial http://logs.openstack.org/37/390937/1/check/gate-packstack-python27-ubuntu-xenial/5a227bc/ : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"43ad252a2059dee28dda4c72325248d7744ffc9a","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-10-27 13:18:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6239a4ea7be021b4fcb8a9a76ca0c07f8c23903c","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2016-10-28 07:29:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"771de57077c26a9c5d17958baf5b9d4be55c8578","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-28 07:30:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"6f19513e2ab2384651d0bc1ab824133fc7ea0c5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-28 09:25:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-packstack-integration-scenario001-tempest-centos-7 http://logs.openstack.org/37/390937/1/gate/gate-packstack-integration-scenario001-tempest-centos-7/f6c50d7/ : SUCCESS in 20m 48s\n- gate-packstack-integration-scenario002-tempest-centos-7 http://logs.openstack.org/37/390937/1/gate/gate-packstack-integration-scenario002-tempest-centos-7/4038a7b/ : SUCCESS in 17m 43s\n- gate-packstack-integration-scenario003-tempest-centos-7 http://logs.openstack.org/37/390937/1/gate/gate-packstack-integration-scenario003-tempest-centos-7/845053e/ : SUCCESS in 21m 51s\n- gate-packstack-pep8-ubuntu-xenial http://logs.openstack.org/37/390937/1/gate/gate-packstack-pep8-ubuntu-xenial/b02ca8c/ : SUCCESS in 1m 22s\n- gate-packstack-python27-ubuntu-xenial http://logs.openstack.org/37/390937/1/gate/gate-packstack-python27-ubuntu-xenial/bd97ceb/ : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"b2d0a91ac9a65e5dec41926615208a52d32d34ae","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-10-28 10:34:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"53f7c843e47107b1260fcdc5a3331857c08bbc06","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2016-10-28 10:36:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d9961ec54613bea577180c91158f66e9b340f041","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-28 11:03:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-packstack-integration-scenario001-tempest-centos-7 http://logs.openstack.org/37/390937/2/check/gate-packstack-integration-scenario001-tempest-centos-7/75fc95f/ : SUCCESS in 21m 29s\n- gate-packstack-integration-scenario002-tempest-centos-7 http://logs.openstack.org/37/390937/2/check/gate-packstack-integration-scenario002-tempest-centos-7/c3f5cbf/ : SUCCESS in 17m 31s\n- gate-packstack-integration-scenario003-tempest-centos-7 http://logs.openstack.org/37/390937/2/check/gate-packstack-integration-scenario003-tempest-centos-7/2ad7ae3/ : SUCCESS in 25m 06s\n- gate-packstack-pep8-ubuntu-xenial http://logs.openstack.org/37/390937/2/check/gate-packstack-pep8-ubuntu-xenial/c9b3dad/ : SUCCESS in 56s\n- gate-packstack-python27-ubuntu-xenial http://logs.openstack.org/37/390937/2/check/gate-packstack-python27-ubuntu-xenial/0dff3a8/ : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2},{"id":"c30246d98918d778a680ea993d30bfb3e5ddb194","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-28 11:03:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"b60c6a898d6f51f277f31c2c8acc289f2a3c1561","date":"2016-10-28 11:26:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"8639cb81db023b6af6e7cec26edf2bc8b34051b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-28 11:26:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-packstack-integration-scenario001-tempest-centos-7 http://logs.openstack.org/37/390937/2/gate/gate-packstack-integration-scenario001-tempest-centos-7/45f9ff8/ : SUCCESS in 18m 56s\n- gate-packstack-integration-scenario002-tempest-centos-7 http://logs.openstack.org/37/390937/2/gate/gate-packstack-integration-scenario002-tempest-centos-7/fc022e5/ : SUCCESS in 17m 47s\n- gate-packstack-integration-scenario003-tempest-centos-7 http://logs.openstack.org/37/390937/2/gate/gate-packstack-integration-scenario003-tempest-centos-7/fc31169/ : SUCCESS in 21m 37s\n- gate-packstack-pep8-ubuntu-xenial http://logs.openstack.org/37/390937/2/gate/gate-packstack-pep8-ubuntu-xenial/82eab4d/ : SUCCESS in 1m 26s\n- gate-packstack-python27-ubuntu-xenial http://logs.openstack.org/37/390937/2/gate/gate-packstack-python27-ubuntu-xenial/a6a2226/ : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd07e91b13263641c3a32017434b3c401f322b86","revisions":{"f6cada13c44d12e2d88f2c84a69feda42be54b7c":{"kind":"REWORK","_number":1,"created":"2016-10-26 17:14:35.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/37/390937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/390937/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/390937/1"}}},"commit":{"parents":[{"commit":"3a7e88c80d0c8935b0eb964567e774fc6ba370c2","subject":"Change default Identity API from v2.0 to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3a7e88c80d0c8935b0eb964567e774fc6ba370c2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2016-10-26 17:12:44.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2016-10-26 17:12:44.000000000","tz":-360},"subject":"Support puppetlabs hiera.yaml update","message":"Support puppetlabs hiera.yaml update\n\nPreviously the code to update the hiera data location for packstack\nassumed /etc/puppet/hiera.yaml which is no longer used in newer versions\nof puppet provided by puppetlabs. This change updates the prescript with\ntwo things. The first it skips creating the symlink to /etc/hiera.yaml\nif /etc/hiera.yaml does not exist. And second it uses the puppet config\ncommand to get the hiera.yaml location to update.\n\nChange-Id: Icd931f36fc14fefbcd43a266f0cd538074723adf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f6cada13c44d12e2d88f2c84a69feda42be54b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f6cada13c44d12e2d88f2c84a69feda42be54b7c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3a7e88c80d0c8935b0eb964567e774fc6ba370c2","is_merged_in_target_branch":true,"change_id":"I0f8519a9994187fd9dca402a1171347a5bb67d8c","change_number":386976,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fd07e91b13263641c3a32017434b3c401f322b86":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-10-28 10:34:15.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/37/390937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/390937/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/390937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/390937/2"}}},"commit":{"parents":[{"commit":"3c005f9260fdce0de9bf2efeb56c3fee2f05fcb9","subject":"Merge \"Switch to ceilometer polling agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3c005f9260fdce0de9bf2efeb56c3fee2f05fcb9"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2016-10-26 17:12:44.000000000","tz":-360},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2016-10-28 10:34:04.000000000","tz":120},"subject":"Support puppetlabs hiera.yaml update","message":"Support puppetlabs hiera.yaml update\n\nPreviously the code to update the hiera data location for packstack\nassumed /etc/puppet/hiera.yaml which is no longer used in newer versions\nof puppet provided by puppetlabs. This change updates the prescript with\ntwo things. The first it skips creating the symlink to /etc/hiera.yaml\nif /etc/hiera.yaml does not exist. And second it uses the puppet config\ncommand to get the hiera.yaml location to update.\n\nChange-Id: Icd931f36fc14fefbcd43a266f0cd538074723adf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fd07e91b13263641c3a32017434b3c401f322b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fd07e91b13263641c3a32017434b3c401f322b86"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3c005f9260fdce0de9bf2efeb56c3fee2f05fcb9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
