)]}'
{"id":"x%2Fpackstack~41614","triplet_id":"x%2Fpackstack~grizzly~I7d1748b01600ee2ce968e2485d6de57b0b95417e","project":"x/packstack","branch":"grizzly","topic":"rawhide","hashtags":[],"change_id":"I7d1748b01600ee2ce968e2485d6de57b0b95417e","subject":"Fix Rawhide support","status":"MERGED","created":"2013-08-13 07:34:44.000000000","updated":"2013-08-20 15:53:22.000000000","submitted":"2013-08-20 15:53:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2654326bb0de5b3d813f6add391e7367d0b22164","_number":41614,"virtual_id_number":41614,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-08-20 15:53:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-08-15 11:05:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},{"value":2,"date":"2013-08-20 15:47:35.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-08-14 09:46:44.000000000","_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},{"value":1,"date":"2013-08-20 15:47:35.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-08-14 09:46:44.000000000","updated_by":{"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},"reviewer":{"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},"state":"REVIEWER"},{"updated":"2013-08-20 15:47:35.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2013-08-20 15:53:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f5146bbf59f33f6630ea0fefd330d3af55a31b86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-13 07:36:52.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/14/41614/1/check/gate-packstack-pep8/b58057a : SUCCESS in 7s\n- http://logs.openstack.org/14/41614/1/check/gate-packstack-python26/a8cf6b5 : SUCCESS in 52s\n- http://logs.openstack.org/14/41614/1/check/gate-packstack-python27/a23a2c5 : SUCCESS in 17s\n","accounts_in_message":[],"_revision_number":1},{"id":"97338169cc233ee5669b797ac4272978ee706b30","author":{"_account_id":6807,"name":"Sandro Mathys","email":"sandro@midokura.com","username":"red-trela"},"date":"2013-08-15 11:05:56.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6900c0bb5b68148f9a562f56f1304014d4abb397","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2013-08-20 15:47:35.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"b3a2019bb89f1d77f11c9b3b12335c1bd1620449","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-20 15:49:05.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"b7eb53cc53d149867f1d9638548b9a2cecc73694","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-20 15:53:22.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- gate-packstack-pep8 http://logs.openstack.org/14/41614/1/gate/gate-packstack-pep8/ed2fb26 : SUCCESS in 8s\n- gate-packstack-python26 http://logs.openstack.org/14/41614/1/gate/gate-packstack-python26/d8a4fc4 : SUCCESS in 1m 22s\n- gate-packstack-python27 http://logs.openstack.org/14/41614/1/gate/gate-packstack-python27/acd5eca : SUCCESS in 16s\n","accounts_in_message":[],"_revision_number":1},{"id":"c9fafd0f58aca8e96245bf602c6821f1b2ade7dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-20 15:53:22.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dcdafd8500e7d69902c9b1d8aec633dbe2c72ef","revisions":{"8dcdafd8500e7d69902c9b1d8aec633dbe2c72ef":{"kind":"REWORK","_number":1,"created":"2013-08-13 07:34:44.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/14/41614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/41614/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/41614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/41614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/41614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/41614/1"}}},"commit":{"parents":[{"commit":"2ed0a8590a4b761c81123c01dfe321159b2abb5d","subject":"Allow tempest repo uri and revision configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2ed0a8590a4b761c81123c01dfe321159b2abb5d"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2013-08-13 06:34:55.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2013-08-13 06:34:55.000000000","tz":120},"subject":"Fix Rawhide support","message":"Fix Rawhide support\n\nChange-Id: I7d1748b01600ee2ce968e2485d6de57b0b95417e\nFixes: rhbz#995872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8dcdafd8500e7d69902c9b1d8aec633dbe2c72ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8dcdafd8500e7d69902c9b1d8aec633dbe2c72ef"}]},"branch":"refs/heads/grizzly"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
