)]}'
{"id":"x%2Fpackstack~647520","triplet_id":"x%2Fpackstack~stable%2Fstein~I7593afe60e1dbde9c713aaa2937997c556707be1","project":"x/packstack","branch":"stable/stein","topic":"647520","hashtags":[],"change_id":"I7593afe60e1dbde9c713aaa2937997c556707be1","subject":"Enable stable/stein reviews","status":"MERGED","created":"2019-03-25 15:59:01.000000000","updated":"2019-04-30 09:48:35.000000000","submitted":"2019-04-30 09:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"647520-1556617715003-47422c5d","meta_rev_id":"0e1d91b0ae420e51867200af7fa4301853d670cc","_number":647520,"virtual_id_number":647520,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-04-30 09:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-04-30 08:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-04-30 08:26:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-27 05:51:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-04-30 08:26:22.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-04-30 09:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcb901c638cc2f2eb33215d3dcce25f4af24fcde","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-03-25 15:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb86f8793a07a6f2f134f8fe90c9a49edbc3cca9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 17:02:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/1/check/openstack-tox-pep8/32443c4/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/1/check/openstack-tox-py27/7f4b6da/ : SUCCESS in 4m 34s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/1/check/packstack-integration-scenario001-tempest/cf1733a/ : SUCCESS in 41m 44s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/1/check/packstack-integration-scenario002-tempest/5f3bab2/ : SUCCESS in 41m 17s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/1/check/packstack-integration-scenario003-tempest/102799d/ : SUCCESS in 51m 36s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/1/check/packstack-multinode-scenario002-tempest/0b718b0/ : SUCCESS in 51m 21s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/20/647520/1/check/packstack-fedora-integration-scenario-py3-tempest/236e6fe/ : SUCCESS in 42m 50s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/20/647520/1/check/openstack-tox-py36/d76244d/ : SUCCESS in 2m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5921c2d56f4ddfd2b2c7df5f9aa1d2b0209aea47","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-03-26 11:32:47.000000000","message":"Patch Set 1:\n\nNeed to switch fedora jobs to consume stein repos https://github.com/openstack/packstack/blob/stable/stein/.zuul.yaml#L49","accounts_in_message":[],"_revision_number":1},{"id":"ea888b4463e72cf56ce046d978d8b52259aba25c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-03-26 12:09:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90b8e9c0dffbfe0e39892f07c5bdee6d93ba8a47","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-03-26 12:10:16.000000000","message":"Patch Set 2:\n\n\u003e Need to switch fedora jobs to consume stein repos https://github.com/openstack/packstack/blob/stable/stein/.zuul.yaml#L49\n\nFixed, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"492d21e6caa33e924ac7b9d7ae6c97d403ca4c06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 13:06:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/2/check/openstack-tox-pep8/2b99ace/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/2/check/openstack-tox-py27/a0d7555/ : SUCCESS in 3m 08s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/2/check/packstack-integration-scenario001-tempest/3a376f7/ : SUCCESS in 42m 47s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/2/check/packstack-integration-scenario002-tempest/95228cc/ : SUCCESS in 45m 35s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/2/check/packstack-integration-scenario003-tempest/f75129d/ : SUCCESS in 48m 28s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/2/check/packstack-multinode-scenario002-tempest/2ed5006/ : SUCCESS in 50m 33s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/20/647520/2/check/packstack-fedora-integration-scenario-py3-tempest/7d7aadc/ : POST_FAILURE in 9m 00s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/20/647520/2/check/openstack-tox-py36/fb73171/ : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"173f58e7069c09cbb704fcd2a856ea40a401afe8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-03-26 17:11:03.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nsome fixes inline","accounts_in_message":[],"_revision_number":2},{"id":"c456e39bc5ef9ebe21707b7ea60a600243b8b922","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-03-26 17:30:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1e7dc7018034d54308c5fe0e0397043fdc7afcb3","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-03-26 17:31:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a4e0fb14c71139ba263f34453becb93edaf8424","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 18:50:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/3/check/openstack-tox-pep8/8185f80/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/3/check/openstack-tox-py27/2b7d1ea/ : SUCCESS in 5m 08s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/3/check/packstack-integration-scenario001-tempest/315b752/ : SUCCESS in 49m 32s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/3/check/packstack-integration-scenario002-tempest/d74a939/ : SUCCESS in 42m 37s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/3/check/packstack-integration-scenario003-tempest/3af78e4/ : SUCCESS in 44m 37s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/3/check/packstack-multinode-scenario002-tempest/01a02c4/ : SUCCESS in 48m 16s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/20/647520/3/check/packstack-fedora-integration-scenario-py3-tempest/d80c825/ : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/20/647520/3/check/openstack-tox-py36/925be4f/ : SUCCESS in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4fb93ff5a63057688bc3a8f582f35abc3d816e4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-03-27 05:51:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c25f7fe56caf4008c1f1cc6d4827d18165a4b3e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-03-27 05:51:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3f8bf5942b9602a6594d753ed53d10f0de1c84dc","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-04-01 18:05:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bef916d80cad311702d37e6fcfa743486a418234","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 16:41:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1897852d184731f51f4b80a2bccf4f82517ca194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 17:43:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/4/check/openstack-tox-pep8/03a1fb9/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/4/check/openstack-tox-py27/3ce036e/ : SUCCESS in 2m 49s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario001-tempest/c603a49/ : SUCCESS in 41m 30s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario002-tempest/50b8760/ : SUCCESS in 41m 12s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario003-tempest/095e8aa/ : SUCCESS in 47m 46s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/4/check/packstack-multinode-scenario002-tempest/dac205b/ : SUCCESS in 48m 14s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/20/647520/4/check/packstack-fedora-integration-scenario-py3-tempest/d3f229a/ : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/20/647520/4/check/openstack-tox-py36/8fea07a/ : SUCCESS in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"166a3126ac76640dda846dce73d1c300199b4bbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 18:58:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/4/check/openstack-tox-pep8/740b777/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/4/check/openstack-tox-py27/46cd38f/ : SUCCESS in 3m 09s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario001-tempest/9872b61/ : SUCCESS in 43m 15s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario002-tempest/8180adc/ : SUCCESS in 39m 22s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/4/check/packstack-integration-scenario003-tempest/c01d594/ : SUCCESS in 46m 46s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/4/check/packstack-multinode-scenario002-tempest/691ec68/ : SUCCESS in 45m 08s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/20/647520/4/check/packstack-fedora-integration-scenario-py3-tempest/b377ca3/ : SUCCESS in 48m 59s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/20/647520/4/check/openstack-tox-py36/12c9dec/ : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bba52ea42b218d5392a55edc9536245b7e02b4b3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-04-30 08:26:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a74de97899c4f7bb3afc72a7f4cb904518869510","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 08:26:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5c2bbb392ff1581e962b0fcd50978bc2360769a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 09:48:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/647520/4/gate/openstack-tox-pep8/66937f6/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/20/647520/4/gate/openstack-tox-py27/516a642/ : SUCCESS in 4m 39s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/20/647520/4/gate/packstack-integration-scenario001-tempest/d2717d3/ : SUCCESS in 45m 17s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/20/647520/4/gate/packstack-integration-scenario002-tempest/cbb0fd8/ : SUCCESS in 44m 24s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/20/647520/4/gate/packstack-integration-scenario003-tempest/24cfc76/ : SUCCESS in 46m 42s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/20/647520/4/gate/packstack-multinode-scenario002-tempest/fe8e6c6/ : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":4},{"id":"521147f1bd57ac6ee7201f1ce25e5a8663a82b28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 09:48:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e9242f14afb4fc5ec242910bf6639fb5da9fa832","revisions":{"0adf5907a4dcc128768a4fadc3507d344f325315":{"kind":"REWORK","_number":1,"created":"2019-03-25 15:59:01.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/20/647520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/647520/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/647520/1"}}},"commit":{"parents":[{"commit":"d781240f86d724cea290bde4a24d91f9b264ad48","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d781240f86d724cea290bde4a24d91f9b264ad48"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-25 15:57:50.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-25 15:57:50.000000000","tz":60},"subject":"Enable stable/stein reviews","message":"Enable stable/stein reviews\n\nWe need to update .gitreview to use the stable/stein branch, and\nthe different sources to use stein-based content, instead of master.\n\nChange-Id: I7593afe60e1dbde9c713aaa2937997c556707be1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0adf5907a4dcc128768a4fadc3507d344f325315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0adf5907a4dcc128768a4fadc3507d344f325315"}]},"branch":"refs/heads/stable/stein"},"6aeec38447f64b9e47b92b43dcebbd4678074a47":{"kind":"REWORK","_number":2,"created":"2019-03-26 12:09:58.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/20/647520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/647520/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/647520/2"}}},"commit":{"parents":[{"commit":"d781240f86d724cea290bde4a24d91f9b264ad48","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d781240f86d724cea290bde4a24d91f9b264ad48"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-25 15:57:50.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-26 12:09:51.000000000","tz":60},"subject":"Enable stable/stein reviews","message":"Enable stable/stein reviews\n\nWe need to update .gitreview to use the stable/stein branch, and\nthe different sources to use stein-based content, instead of master.\n\nChange-Id: I7593afe60e1dbde9c713aaa2937997c556707be1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6aeec38447f64b9e47b92b43dcebbd4678074a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6aeec38447f64b9e47b92b43dcebbd4678074a47"}]},"branch":"refs/heads/stable/stein"},"f50baed03aa2f230ee2fe762f7d3d91e703f208c":{"kind":"REWORK","_number":3,"created":"2019-03-26 17:30:36.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/20/647520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/647520/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/647520/3"}}},"commit":{"parents":[{"commit":"d781240f86d724cea290bde4a24d91f9b264ad48","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d781240f86d724cea290bde4a24d91f9b264ad48"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-25 15:57:50.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-03-26 17:30:34.000000000","tz":0},"subject":"Enable stable/stein reviews","message":"Enable stable/stein reviews\n\nWe need to update .gitreview to use the stable/stein branch, and\nthe different sources to use stein-based content, instead of master.\n\nChange-Id: I7593afe60e1dbde9c713aaa2937997c556707be1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f50baed03aa2f230ee2fe762f7d3d91e703f208c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f50baed03aa2f230ee2fe762f7d3d91e703f208c"}]},"branch":"refs/heads/stable/stein"},"e9242f14afb4fc5ec242910bf6639fb5da9fa832":{"kind":"REWORK","_number":4,"created":"2019-04-29 16:41:58.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/20/647520/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/647520/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/647520/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/647520/4"}}},"commit":{"parents":[{"commit":"269243890d8724bfa2212dfd7d3957050fd5bfce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/269243890d8724bfa2212dfd7d3957050fd5bfce"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-03-25 15:57:50.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 16:41:56.000000000","tz":120},"subject":"Enable stable/stein reviews","message":"Enable stable/stein reviews\n\nWe need to update .gitreview to use the stable/stein branch, and\nthe different sources to use stein-based content, instead of master.\n\nChange-Id: I7593afe60e1dbde9c713aaa2937997c556707be1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e9242f14afb4fc5ec242910bf6639fb5da9fa832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e9242f14afb4fc5ec242910bf6639fb5da9fa832"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
