)]}'
{"id":"x%2Fpackstack~656293","triplet_id":"x%2Fpackstack~master~I2978015a32101de7a27e13152d89b3802225b128","project":"x/packstack","branch":"master","topic":"placement","hashtags":[],"change_id":"I2978015a32101de7a27e13152d89b3802225b128","subject":"Move placement to extracted service","status":"MERGED","created":"2019-04-29 11:30:34.000000000","updated":"2019-04-30 09:48:33.000000000","submitted":"2019-04-30 09:48:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656293-1556617713995-0411a076","meta_rev_id":"a6180bdb6d111141335fc1400f933417075cbdd8","_number":656293,"virtual_id_number":656293,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-04-30 09:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"recommended":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":1,"date":"2019-04-30 07:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":1,"date":"2019-04-29 16:26:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-04-29 19:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2019-04-29 16:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2019-04-30 08:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 16:26:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-04-29 19:28:02.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-04-30 07:08:52.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2019-04-30 08:25:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-04-30 09:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c3f6fe21d8932e7d40c0f6170a8d9889fba0655","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 11:30:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e06c17bc8caf83f412d2f6a7e1552ad7fd95aed9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 12:34:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/93/656293/1/check/openstack-tox-pep8/8804426/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/93/656293/1/check/openstack-tox-py27/effafe2/ : SUCCESS in 3m 40s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/93/656293/1/check/packstack-integration-scenario001-tempest/651bb89/ : FAILURE in 37m 42s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/93/656293/1/check/packstack-integration-scenario002-tempest/6112d2c/ : FAILURE in 44m 16s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/93/656293/1/check/packstack-integration-scenario003-tempest/ed99557/ : FAILURE in 48m 13s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/93/656293/1/check/packstack-multinode-scenario002-tempest/02a99c3/ : FAILURE in 50m 22s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/93/656293/1/check/packstack-fedora-integration-scenario-py3-tempest/7408b4a/ : FAILURE in 40m 48s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/93/656293/1/check/openstack-tox-py36/68d878b/ : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5df6d811f23b633b4e80a0370af43720e99969b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-04-29 12:36:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThanks Javier! You\u0027ve just missed that some uses of ::nova::placement need to remain to update nova.conf so we can find and use the Placement endpoint but aside from that it looks good.","accounts_in_message":[],"_revision_number":1},{"id":"abf7fc252fb358a2079af5dc438bf59e0736f070","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 12:47:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e985a3609e567d296a49148f30d5a3e0ed119fc","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 12:48:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94a9d7e39f4dd67412b14c6ef22f3745aed7a512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 13:40:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/93/656293/2/check/openstack-tox-pep8/d0b0785/ : SUCCESS in 2m 16s\n- openstack-tox-py27 http://logs.openstack.org/93/656293/2/check/openstack-tox-py27/3fbba99/ : SUCCESS in 3m 56s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/93/656293/2/check/packstack-integration-scenario001-tempest/ca1c3bf/ : FAILURE in 39m 10s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/93/656293/2/check/packstack-integration-scenario002-tempest/9f9cb43/ : FAILURE in 40m 47s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/93/656293/2/check/packstack-integration-scenario003-tempest/8637d66/ : FAILURE in 45m 39s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/93/656293/2/check/packstack-multinode-scenario002-tempest/e0661dc/ : FAILURE in 41m 09s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/93/656293/2/check/packstack-fedora-integration-scenario-py3-tempest/2770cc6/ : FAILURE in 34m 38s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/93/656293/2/check/openstack-tox-py36/2efa6b5/ : SUCCESS in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc89d390e7b2aab7b4c09d70f76579b5ee30b75d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 14:47:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"110dd3ec53068d8f57377797eef1f3b298fcb3b9","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-04-29 14:56:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3a18f4815253d922a1bda542d78b517f551ddf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:59:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/93/656293/4/check/openstack-tox-pep8/a5ead9e/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/93/656293/4/check/openstack-tox-py27/edc09f4/ : SUCCESS in 2m 55s\n- build-openstack-releasenotes http://logs.openstack.org/93/656293/4/check/build-openstack-releasenotes/eb3792d/html/ : SUCCESS in 3m 53s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/93/656293/4/check/packstack-integration-scenario001-tempest/5fc0e9d/ : SUCCESS in 44m 53s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/93/656293/4/check/packstack-integration-scenario002-tempest/9cad672/ : SUCCESS in 56m 07s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/93/656293/4/check/packstack-integration-scenario003-tempest/a1a28bd/ : SUCCESS in 47m 55s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/93/656293/4/check/packstack-multinode-scenario002-tempest/7c0c041/ : SUCCESS in 46m 55s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/93/656293/4/check/packstack-fedora-integration-scenario-py3-tempest/1b7ed17/ : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/93/656293/4/check/openstack-tox-py36/dde6974/ : SUCCESS in 2m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"56d0f8844d6e8d8fb1f4d003d600a9b17bf2b9d5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-04-29 16:10:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"db65d0ac30a86f27ee381ba654e035de69360bc0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-04-29 16:26:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4ac700045c9d6dfe9ca733677a294ac24f2b019f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-04-29 19:28:02.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM! Apologies again for letting this slip through.","accounts_in_message":[],"_revision_number":4},{"id":"cf6e3c7f38151fa7d09ca62ee9bc09329a131e03","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-04-30 07:08:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"815fdab663510965b77e8afc514b14fc2113eb07","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-04-30 08:25:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"891cd3ecaeadd48f7cb329efd6f2ca4e9f45dba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 08:25:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23ced3a9e1fdaab1650e5f24e130b0d413062f3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 09:48:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/93/656293/4/gate/openstack-tox-pep8/612866a/ : SUCCESS in 2m 18s\n- openstack-tox-py27 http://logs.openstack.org/93/656293/4/gate/openstack-tox-py27/7d91232/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/93/656293/4/gate/build-openstack-releasenotes/f55f861/html/ : SUCCESS in 2m 30s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/93/656293/4/gate/packstack-integration-scenario001-tempest/a82d156/ : SUCCESS in 44m 40s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/93/656293/4/gate/packstack-integration-scenario002-tempest/a4e110a/ : SUCCESS in 45m 23s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/93/656293/4/gate/packstack-integration-scenario003-tempest/c627e1d/ : SUCCESS in 52m 14s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/93/656293/4/gate/packstack-multinode-scenario002-tempest/2b19acd/ : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d031833ae47883e8b3ab4af4c18c59a2b9ba89c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 09:48:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7afb1db0439b778f0340b890b5813429bfde0496","revisions":{"de7cb61accef362145d03cad401836649df15612":{"kind":"REWORK","_number":1,"created":"2019-04-29 11:30:34.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/93/656293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/656293/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/656293/1"}}},"commit":{"parents":[{"commit":"247b56d07c22bfed3cb17ea8c38afa6222e13e9c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/247b56d07c22bfed3cb17ea8c38afa6222e13e9c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 11:03:04.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 11:30:31.000000000","tz":120},"subject":"[WIP] Move placement to extracted service","message":"[WIP] Move placement to extracted service\n\nMissing:\n\n- Release notes\n\nChange-Id: I2978015a32101de7a27e13152d89b3802225b128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/de7cb61accef362145d03cad401836649df15612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/de7cb61accef362145d03cad401836649df15612"}]},"branch":"refs/heads/master"},"3a7cd1b1300ef5b7410a110e92cfeb014afec7a7":{"kind":"REWORK","_number":2,"created":"2019-04-29 12:48:33.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/93/656293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/656293/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/656293/2"}}},"commit":{"parents":[{"commit":"247b56d07c22bfed3cb17ea8c38afa6222e13e9c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/247b56d07c22bfed3cb17ea8c38afa6222e13e9c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 11:03:04.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 12:48:26.000000000","tz":120},"subject":"[WIP] Move placement to extracted service","message":"[WIP] Move placement to extracted service\n\nMissing:\n\n- Release notes\n\nChange-Id: I2978015a32101de7a27e13152d89b3802225b128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3a7cd1b1300ef5b7410a110e92cfeb014afec7a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3a7cd1b1300ef5b7410a110e92cfeb014afec7a7"}]},"branch":"refs/heads/master"},"485c1f9e9eafc9a5260ccdccddbd6d78d87184de":{"kind":"REWORK","_number":3,"created":"2019-04-29 14:47:40.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/93/656293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/656293/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/656293/3"}}},"commit":{"parents":[{"commit":"247b56d07c22bfed3cb17ea8c38afa6222e13e9c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/247b56d07c22bfed3cb17ea8c38afa6222e13e9c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 11:03:04.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 14:44:12.000000000","tz":120},"subject":"Move placement to extracted service","message":"Move placement to extracted service\n\nPuppet-nova has removed the support for the integrated service in [1],\nso Packstack has to move along and start supporting the extracted\nservice.\n\n[1] - https://review.opendev.org/638657\n\nChange-Id: I2978015a32101de7a27e13152d89b3802225b128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/485c1f9e9eafc9a5260ccdccddbd6d78d87184de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/485c1f9e9eafc9a5260ccdccddbd6d78d87184de"}]},"branch":"refs/heads/master"},"7afb1db0439b778f0340b890b5813429bfde0496":{"kind":"REWORK","_number":4,"created":"2019-04-29 14:56:11.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/93/656293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/656293/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/656293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/656293/4"}}},"commit":{"parents":[{"commit":"247b56d07c22bfed3cb17ea8c38afa6222e13e9c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/247b56d07c22bfed3cb17ea8c38afa6222e13e9c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 11:03:04.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-04-29 14:56:03.000000000","tz":120},"subject":"Move placement to extracted service","message":"Move placement to extracted service\n\nPuppet-nova has removed the support for the integrated service in [1],\nso Packstack has to move along and start supporting the extracted\nservice.\n\n[1] - https://review.opendev.org/638657\n\nChange-Id: I2978015a32101de7a27e13152d89b3802225b128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7afb1db0439b778f0340b890b5813429bfde0496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7afb1db0439b778f0340b890b5813429bfde0496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
