)]}'
{"id":"x%2Fpackstack~658832","triplet_id":"x%2Fpackstack~master~I055ab99e48e018d041a2aef7c8998f72ebd4ff5e","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I055ab99e48e018d041a2aef7c8998f72ebd4ff5e","subject":"Pin sphinx to \u003c2.0.0","status":"MERGED","created":"2019-05-13 14:28:07.000000000","updated":"2019-06-05 09:04:38.000000000","submitted":"2019-05-16 13:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658832-1558013311689-cf8f5779","meta_rev_id":"e7a60c8992e82e0069fbc8187ecb163bb72c5656","_number":658832,"virtual_id_number":658832,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-05-16 13:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-05-14 07:06:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-05-16 12:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-13 15:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-05-16 12:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 15:32:43.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-05-14 07:06:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-05-16 12:33:49.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-05-16 13:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"392b230fcf5640e4d4c2fa9b0653c45d15eadcc9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-13 14:28:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5351a742f4d7a87119c29c4613f90d9a8209ec9c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-05-13 15:32:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"126215c05b861ffd5bf616594526cce80d0cc33f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:33:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/32/658832/1/check/openstack-tox-pep8/aa04235/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/32/658832/1/check/openstack-tox-py27/301e15d/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes http://logs.openstack.org/32/658832/1/check/build-openstack-releasenotes/c0d0af7/html/ : SUCCESS in 2m 23s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/32/658832/1/check/packstack-integration-scenario001-tempest/afed314/ : SUCCESS in 48m 42s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/32/658832/1/check/packstack-integration-scenario002-tempest/fc5d7fd/ : SUCCESS in 46m 27s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/32/658832/1/check/packstack-integration-scenario003-tempest/1aa4c67/ : SUCCESS in 58m 09s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/32/658832/1/check/packstack-multinode-scenario002-tempest/a51a952/ : SUCCESS in 52m 16s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/32/658832/1/check/packstack-fedora-integration-scenario-py3-tempest/0d3e9d3/ : SUCCESS in 38m 48s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/32/658832/1/check/openstack-tox-py36/07d328b/ : SUCCESS in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac50b8faa3ff964c6610c1895e837975eff630fb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-14 07:06:17.000000000","message":"Patch Set 1: Code-Review+1\n\nsphinx should not be needed for tests, so good to seperate out doc requirement, anyway it\u0027s a followup task so let\u0027s do later.","accounts_in_message":[],"_revision_number":1},{"id":"541c943852784574de4fcc7ca10ce8f58fb4694e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-05-16 12:33:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9af63c7829743360256db3a8b6252af8522cdda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 12:34:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77eb51a57bedc811f1709be727812667fd5a3e07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/32/658832/1/gate/openstack-tox-pep8/47a10b2/ : SUCCESS in 2m 26s\n- openstack-tox-py27 http://logs.openstack.org/32/658832/1/gate/openstack-tox-py27/d23bf6e/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/32/658832/1/gate/build-openstack-releasenotes/da1d5f4/html/ : SUCCESS in 2m 10s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/32/658832/1/gate/packstack-integration-scenario001-tempest/c5f76b9/ : SUCCESS in 47m 39s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/32/658832/1/gate/packstack-integration-scenario002-tempest/142bad9/ : SUCCESS in 48m 50s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/32/658832/1/gate/packstack-integration-scenario003-tempest/a1fb4d6/ : SUCCESS in 44m 32s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/32/658832/1/gate/packstack-multinode-scenario002-tempest/a61a1b1/ : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"27dfd6e548b2f2d389e3fe7f69b32d64dae18062","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:28:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d5fa1c635776a7bd3c1cc135d1c099032038bfc","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-05-17 16:07:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b8c44d79760e74cec61604817b61f14b82275411","accounts_in_message":[],"_revision_number":1},{"id":"e77537da5477a4728365bbf888a268581e1b0016","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-05 09:04:24.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit da66dc028a849e4db0e47a9dc3d080bcee4ca871","accounts_in_message":[],"_revision_number":1},{"id":"e7a60c8992e82e0069fbc8187ecb163bb72c5656","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-05 09:04:38.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 315d55ce86f7f2cab05afc62db058156cf3bd1ed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2b1c67d57035416bf04f04dabb4e927216836ff","revisions":{"d2b1c67d57035416bf04f04dabb4e927216836ff":{"kind":"REWORK","_number":1,"created":"2019-05-13 14:28:07.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/32/658832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/658832/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/658832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/658832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/658832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/658832/1"}}},"commit":{"parents":[{"commit":"7afb1db0439b778f0340b890b5813429bfde0496","subject":"Move placement to extracted service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7afb1db0439b778f0340b890b5813429bfde0496"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-13 14:27:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-13 14:27:19.000000000","tz":120},"subject":"Pin sphinx to \u003c2.0.0","message":"Pin sphinx to \u003c2.0.0\n\nThe 2.0.0 release introduces a requirement for using\npython3.5 but we are running this in python 2.7\n\nChange-Id: I055ab99e48e018d041a2aef7c8998f72ebd4ff5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d2b1c67d57035416bf04f04dabb4e927216836ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d2b1c67d57035416bf04f04dabb4e927216836ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
