)]}'
{"id":"x%2Fpackstack~660371","triplet_id":"x%2Fpackstack~master~Id96c234d94764045a1e24afd1964f9b85cd08142","project":"x/packstack","branch":"master","topic":"fix-packstack","hashtags":[],"change_id":"Id96c234d94764045a1e24afd1964f9b85cd08142","subject":"Remove parameters removed from nova::network::neutron","status":"MERGED","created":"2019-05-21 12:41:01.000000000","updated":"2019-05-21 19:02:22.000000000","submitted":"2019-05-21 19:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660371-1558465342321-54bd5aa7","meta_rev_id":"7c27628d43b88ea1fb6fd97d1c82f195958d2c6b","_number":660371,"virtual_id_number":660371,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2019-05-21 16:26:12.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-05-21 19:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-05-21 12:51:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 14:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 14:16:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 12:51:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-05-21 14:16:49.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-05-21 19:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c1c9ecc9f520b7f6d5f20cff78d1330fccfce53","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-05-21 12:41:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f52c477fca923338161041145ab7dd6dfd43bb3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-21 12:51:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8a9980d36e6d0602dc93e71105cb1b5d66392c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:13:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/660371/1/check/openstack-tox-pep8/0cc8498/ : SUCCESS in 2m 30s\n- openstack-tox-py27 http://logs.openstack.org/71/660371/1/check/openstack-tox-py27/af9040a/ : SUCCESS in 2m 51s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario001-tempest/25c4193/ : SUCCESS in 1h 04m 31s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario002-tempest/e16cbf5/ : SUCCESS in 1h 11m 06s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario003-tempest/f46c02c/ : SUCCESS in 1h 18m 29s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/71/660371/1/check/packstack-multinode-scenario002-tempest/e800656/ : SUCCESS in 1h 11m 35s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/71/660371/1/check/packstack-fedora-integration-scenario-py3-tempest/0058f93/ : SUCCESS in 54m 44s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/71/660371/1/check/openstack-tox-py36/579a54a/ : SUCCESS in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"104da753f3ea6caff7fcbdb2bcd1e5637613e6da","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-05-21 14:16:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b9f75232b89f1284fde46acdedad08d335aaf15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:17:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1831d9b8e710f6f5221eaffa1a04efe2bab5e99e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:10:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/71/660371/1/gate/openstack-tox-pep8/f15b6b2/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/71/660371/1/gate/openstack-tox-py27/6e2b177/ : SUCCESS in 2m 43s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario001-tempest/8be7d0a/ : SUCCESS in 53m 11s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario002-tempest/de653e4/ : SUCCESS in 1h 01m 15s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario003-tempest/2736f12/ : FAILURE in 1h 00m 23s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/71/660371/1/gate/packstack-multinode-scenario002-tempest/56c6cd6/ : SUCCESS in 1h 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3e4a1b9bb846fead9e2e722c70bf87811835e3e5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-05-21 16:26:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a794c2970cb2714da636072769b6b5d8f9c3bf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/660371/1/check/openstack-tox-pep8/288873a/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/71/660371/1/check/openstack-tox-py27/60e079b/ : SUCCESS in 2m 49s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario001-tempest/c02639b/ : SUCCESS in 1h 03m 37s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario002-tempest/aedf356/ : SUCCESS in 1h 08m 57s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/71/660371/1/check/packstack-integration-scenario003-tempest/f5c5fda/ : SUCCESS in 57m 23s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/71/660371/1/check/packstack-multinode-scenario002-tempest/c979f3d/ : SUCCESS in 1h 08m 06s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/71/660371/1/check/packstack-fedora-integration-scenario-py3-tempest/f1aa637/ : SUCCESS in 51m 21s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/71/660371/1/check/openstack-tox-py36/e5d4140/ : SUCCESS in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27e9a0326ddaeb1505cbb73cb18e3911e04bd452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:41:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f265399449ea25264d64c623c756480d603b362","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:02:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/660371/1/gate/openstack-tox-pep8/64d4b39/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/71/660371/1/gate/openstack-tox-py27/72b61cb/ : SUCCESS in 3m 01s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario001-tempest/884044d/ : SUCCESS in 1h 06m 14s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario002-tempest/fffa927/ : SUCCESS in 1h 12m 44s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/71/660371/1/gate/packstack-integration-scenario003-tempest/8e98188/ : SUCCESS in 1h 05m 05s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/71/660371/1/gate/packstack-multinode-scenario002-tempest/864ba42/ : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b20f9d30219e28bb842b00f4b0f70bb93b3560c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:02:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7683bb60cd653cce28c3e163b0ba508f912cee52","revisions":{"7683bb60cd653cce28c3e163b0ba508f912cee52":{"kind":"REWORK","_number":1,"created":"2019-05-21 12:41:01.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/71/660371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/71/660371/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/71/660371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/71/660371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/71/660371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/71/660371/1"}}},"commit":{"parents":[{"commit":"09199d74ae1b68063806e58662653ab455a77bde","subject":"Merge \"Fix URL to get the list of projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/09199d74ae1b68063806e58662653ab455a77bde"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-05-21 12:38:17.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-05-21 12:40:56.000000000","tz":120},"subject":"Remove parameters removed from nova::network::neutron","message":"Remove parameters removed from nova::network::neutron\n\nSome neutron parameters have been removed from puppet-nova so we need to\nretire them from packstack. Now neutron_url is fetched from keystone\ncatalog.\n\n[1] https://review.opendev.org/#/c/658400/\n\nChange-Id: Id96c234d94764045a1e24afd1964f9b85cd08142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7683bb60cd653cce28c3e163b0ba508f912cee52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7683bb60cd653cce28c3e163b0ba508f912cee52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
