)]}'
{"id":"x%2Fpackstack~660985","triplet_id":"x%2Fpackstack~master~Iabf8170a6e584dd3bb42c1a5e3a4c2c078c042ac","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Iabf8170a6e584dd3bb42c1a5e3a4c2c078c042ac","subject":"Remove reference to ::nova::compute::neutron","status":"MERGED","created":"2019-05-23 12:44:12.000000000","updated":"2019-05-23 15:34:16.000000000","submitted":"2019-05-23 15:34:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660985-1558625656808-95f0d4b7","meta_rev_id":"d8f29bbc6d4035110339fa7483b661b1c5cbb491","_number":660985,"virtual_id_number":660985,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-05-23 15:34:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":1,"date":"2019-05-23 14:09:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-05-23 14:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-23 14:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-23 14:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 14:06:07.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-05-23 14:09:57.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-05-23 14:11:33.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-05-23 15:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd8082c79a0dfcca06fcc5ee66996d480b5cc957","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-05-23 12:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"980b3aa345d187f08ddded8b3be5c2f7b57872ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:05:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/660985/1/check/openstack-tox-pep8/2cfa7d1/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/85/660985/1/check/openstack-tox-py27/1d97e57/ : SUCCESS in 3m 11s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/85/660985/1/check/packstack-integration-scenario001-tempest/3d414a4/ : SUCCESS in 1h 02m 08s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/85/660985/1/check/packstack-integration-scenario002-tempest/15e65ec/ : SUCCESS in 1h 03m 47s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/85/660985/1/check/packstack-integration-scenario003-tempest/9fc8af5/ : SUCCESS in 1h 09m 05s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/85/660985/1/check/packstack-multinode-scenario002-tempest/70e1191/ : SUCCESS in 57m 02s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/85/660985/1/check/packstack-fedora-integration-scenario-py3-tempest/104037f/ : SUCCESS in 48m 53s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/85/660985/1/check/openstack-tox-py36/42b69a4/ : SUCCESS in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8256c45c0cfa832891dc513b3ac8f81d3224b7a4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-23 14:06:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"68727be5880278c99b4a3f0dd4b072ce9073f0e2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-05-23 14:09:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e839effb36db2b1f1af59010f69a135a7054895f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-05-23 14:11:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c723c5303d7fad84bccb312fa80f1d51dd5c402d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:11:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69cd8fc247f469b1ed176b6ade89a0dcefb6d9ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:34:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d8f29bbc6d4035110339fa7483b661b1c5cbb491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:34:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/660985/1/gate/openstack-tox-pep8/c3fbec2/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/85/660985/1/gate/openstack-tox-py27/b1b496d/ : SUCCESS in 3m 09s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/85/660985/1/gate/packstack-integration-scenario001-tempest/6dba924/ : SUCCESS in 1h 06m 33s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/85/660985/1/gate/packstack-integration-scenario002-tempest/1dbe2f2/ : SUCCESS in 1h 15m 53s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/85/660985/1/gate/packstack-integration-scenario003-tempest/64b7be2/ : SUCCESS in 1h 02m 45s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/85/660985/1/gate/packstack-multinode-scenario002-tempest/8cc2a80/ : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e678051957840c460289848bd8cb83ca78bf15f","revisions":{"9e678051957840c460289848bd8cb83ca78bf15f":{"kind":"REWORK","_number":1,"created":"2019-05-23 12:44:12.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/85/660985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/660985/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/660985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/660985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/660985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/660985/1"}}},"commit":{"parents":[{"commit":"7683bb60cd653cce28c3e163b0ba508f912cee52","subject":"Remove parameters removed from nova::network::neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7683bb60cd653cce28c3e163b0ba508f912cee52"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-05-23 12:35:43.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-05-23 12:35:43.000000000","tz":120},"subject":"Remove reference to ::nova::compute::neutron","message":"Remove reference to ::nova::compute::neutron\n\nThis class has been removed from puppet-nova as it only configures a\nparameter related to nova-network which has been finally removed from\npuppet-nova [1].\n\n[1] https://review.opendev.org/#/c/658403/\n\nChange-Id: Iabf8170a6e584dd3bb42c1a5e3a4c2c078c042ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e678051957840c460289848bd8cb83ca78bf15f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e678051957840c460289848bd8cb83ca78bf15f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
