)]}'
{"id":"x%2Fpackstack~663275","triplet_id":"x%2Fpackstack~stable%2Frocky~I45dc6511ee8874b9719be6816062777f2ce371f0","project":"x/packstack","branch":"stable/rocky","hashtags":[],"change_id":"I45dc6511ee8874b9719be6816062777f2ce371f0","subject":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file","status":"MERGED","created":"2019-06-05 08:50:26.000000000","updated":"2019-06-05 15:14:40.000000000","submitted":"2019-06-05 15:14:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663275-1559747680334-f390cf89","meta_rev_id":"41dd21c907dbaf8012c0b2da9d665581107e16f0","_number":663275,"virtual_id_number":663275,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-05 15:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 12:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 14:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 14:09:07.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-06-05 15:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5642051efc1df01892079b3b3d24e01938d3e2e1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-05 08:50:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"6871a7c32f31f21793de4e31d13930238e85c677","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-05 09:14:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"46b57f62100b8477197055ebbbe130241277edfa","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-05 11:04:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"981e8d3671c7c0fcdd283b3158dcaec7409526ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:02:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/663275/3/check/openstack-tox-pep8/149aed2/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/75/663275/3/check/openstack-tox-py27/ac46629/ : SUCCESS in 3m 09s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/75/663275/3/check/packstack-integration-scenario001-tempest/2e40ac2/ : SUCCESS in 49m 51s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/75/663275/3/check/packstack-integration-scenario002-tempest/b165a2c/ : SUCCESS in 40m 50s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/75/663275/3/check/packstack-integration-scenario003-tempest/a436ffd/ : SUCCESS in 51m 40s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/75/663275/3/check/packstack-multinode-scenario002-tempest/9d1a5e0/ : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":3},{"id":"89bab4cd79c182e7a05142ab7a7189d566200c8e","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-06-05 12:44:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"13dbdb62e80f6304763fa8458265ac2c53631229","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-06-05 14:09:07.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"38b5505d5ab363a39cea07d48abc53702aef29ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:09:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"32d761e8f0f4bb14c89ef8595e77a59d16724e87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:14:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"41dd21c907dbaf8012c0b2da9d665581107e16f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:14:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/663275/3/gate/openstack-tox-pep8/90cc6ba/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/75/663275/3/gate/openstack-tox-py27/904bd13/ : SUCCESS in 2m 56s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/75/663275/3/gate/packstack-integration-scenario001-tempest/cf9cfe4/ : SUCCESS in 57m 02s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/75/663275/3/gate/packstack-integration-scenario002-tempest/efa35b0/ : SUCCESS in 39m 52s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/75/663275/3/gate/packstack-integration-scenario003-tempest/cde8f35/ : SUCCESS in 46m 48s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/75/663275/3/gate/packstack-multinode-scenario002-tempest/e895214/ : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45a43f0a206edf95b54f0364df33f7c7beed50d5","revisions":{"196ff79f731d9976d08c3754889d6ab288bbf755":{"kind":"REWORK","_number":1,"created":"2019-06-05 08:50:26.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/75/663275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/663275/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/663275/1"}}},"commit":{"parents":[{"commit":"ce737f89b1da06f6d9e38a97e26cccbf7fa36143","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ce737f89b1da06f6d9e38a97e26cccbf7fa36143"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-04 16:29:57.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-05 08:50:26.000000000","tz":0},"subject":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file","message":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file\n\nCurrently, OS_REGION_NAME is in keystone_admin rc file only when\nkeystone api is \u003c 3. This is incorrect as region name is also meaningful\nin newer versions and needed in some cases as shown in [1].\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1716434\n\nChange-Id: I45dc6511ee8874b9719be6816062777f2ce371f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/196ff79f731d9976d08c3754889d6ab288bbf755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/196ff79f731d9976d08c3754889d6ab288bbf755"}]},"branch":"refs/heads/stable/rocky"},"e2849d0bbaf66305d1ce658db7f915b0a42dcbd5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-05 09:14:39.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/75/663275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/663275/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/663275/2"}}},"commit":{"parents":[{"commit":"da66dc028a849e4db0e47a9dc3d080bcee4ca871","subject":"Pin sphinx to \u003c2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/da66dc028a849e4db0e47a9dc3d080bcee4ca871"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-04 16:29:57.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-05 09:14:39.000000000","tz":0},"subject":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file","message":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file\n\nCurrently, OS_REGION_NAME is in keystone_admin rc file only when\nkeystone api is \u003c 3. This is incorrect as region name is also meaningful\nin newer versions and needed in some cases as shown in [1].\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1716434\n\nChange-Id: I45dc6511ee8874b9719be6816062777f2ce371f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e2849d0bbaf66305d1ce658db7f915b0a42dcbd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e2849d0bbaf66305d1ce658db7f915b0a42dcbd5"}]},"branch":"refs/heads/stable/rocky"},"45a43f0a206edf95b54f0364df33f7c7beed50d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-05 11:04:23.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/75/663275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/663275/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/663275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/663275/3"}}},"commit":{"parents":[{"commit":"53a73ba0eaa08b9e2d0da1fe8221fc60be709346","subject":"Pin sphinx to \u003c2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/53a73ba0eaa08b9e2d0da1fe8221fc60be709346"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-04 16:29:57.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-06-05 11:04:23.000000000","tz":0},"subject":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file","message":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file\n\nCurrently, OS_REGION_NAME is in keystone_admin rc file only when\nkeystone api is \u003c 3. This is incorrect as region name is also meaningful\nin newer versions and needed in some cases as shown in [1].\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1716434\n\nChange-Id: I45dc6511ee8874b9719be6816062777f2ce371f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45a43f0a206edf95b54f0364df33f7c7beed50d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45a43f0a206edf95b54f0364df33f7c7beed50d5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
