)]}'
{"id":"x%2Fpackstack~668736","triplet_id":"x%2Fpackstack~master~Idf0a0071446d87c5c6215c037f2dab84df96662d","project":"x/packstack","branch":"master","topic":"f29","hashtags":[],"change_id":"Idf0a0071446d87c5c6215c037f2dab84df96662d","subject":"DNM: test patch to use f29 nodeset from devstack","status":"ABANDONED","created":"2019-07-02 18:40:10.000000000","updated":"2019-09-13 18:32:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21b0d1d83bc072e2b1fb941b84f944fb34a6bcb8","_number":668736,"virtual_id_number":668736,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-07-08 20:25:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2019-07-02 18:40:20.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 20:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a2555bbf5346a8279556b92b58a8db366f8d982","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:40:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f0951532b5e0904c901d1ba6a2a6f3ce82e5fcf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:40:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7a8ced0e2d7ce9394b644b1bc49e35bea58e59fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/668736/1/check/openstack-tox-pep8/fa495b4/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/36/668736/1/check/openstack-tox-py27/e204a5f/ : SUCCESS in 3m 17s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario001-tempest/cbb2ed7/ : FAILURE in 1h 53m 56s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario002-tempest/8e1dcc0/ : SUCCESS in 1h 04m 58s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario003-tempest/0b73a9f/ : SUCCESS in 57m 50s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/36/668736/1/check/packstack-multinode-scenario002-tempest/3b77db1/ : SUCCESS in 1h 17m 15s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/36/668736/1/check/packstack-fedora-integration-scenario-py3-tempest/ca33500/ : RETRY_LIMIT in 3m 35s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/36/668736/1/check/openstack-tox-py36/4912047/ : SUCCESS in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d245aadaa9e731069aa8129c9eee7dd4c373d1d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-08 19:04:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"228ab3d35072c9b8ace18ed26396d53610ab16b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:25:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/668736/1/check/openstack-tox-pep8/ca332f2/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/36/668736/1/check/openstack-tox-py27/aed3dff/ : SUCCESS in 3m 15s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario001-tempest/fc7075d/ : SUCCESS in 1h 14m 36s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario002-tempest/d6b61cf/ : SUCCESS in 53m 58s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/36/668736/1/check/packstack-integration-scenario003-tempest/e6d7a32/ : SUCCESS in 59m 09s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/36/668736/1/check/packstack-multinode-scenario002-tempest/57ff09d/ : SUCCESS in 1h 11m 13s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/36/668736/1/check/packstack-fedora-integration-scenario-py3-tempest/fd78647/ : RETRY_LIMIT in 4m 11s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/36/668736/1/check/openstack-tox-py36/81b2ad0/ : SUCCESS in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49b30d6f916cc59e52cb12d2fac66bba43d53a64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-13 18:32:06.000000000","message":"Abandoned\n\nWas just a test patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cc90da5a208aae0cfc87483cdc6420704c06547","revisions":{"7cc90da5a208aae0cfc87483cdc6420704c06547":{"kind":"REWORK","_number":1,"created":"2019-07-02 18:40:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/668736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/36/668736/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/36/668736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/36/668736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/36/668736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/36/668736/1"}}},"commit":{"parents":[{"commit":"f8a0048df2f31903ab08a801c75d754439a63475","subject":"Merge \"Fixes typo in neutron_350.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8a0048df2f31903ab08a801c75d754439a63475"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:39:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:40:09.000000000","tz":-240},"subject":"DNM: test patch to use f29 nodeset from devstack","message":"DNM: test patch to use f29 nodeset from devstack\n\nDepends-on: https://review.opendev.org/#/c/662529/\nChange-Id: Idf0a0071446d87c5c6215c037f2dab84df96662d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7cc90da5a208aae0cfc87483cdc6420704c06547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7cc90da5a208aae0cfc87483cdc6420704c06547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
