)]}'
{"id":"x%2Fpackstack~672692","triplet_id":"x%2Fpackstack~master~I1e53bbdcccb6402d343c98abf2696d452717d1a9","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I1e53bbdcccb6402d343c98abf2696d452717d1a9","subject":"Force qemu emulation in CI and add dmidecode command to logs","status":"MERGED","created":"2019-07-25 10:59:45.000000000","updated":"2019-08-13 07:44:55.000000000","submitted":"2019-07-26 12:08:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672692-1564142887285-a68571c6","meta_rev_id":"fb64fce18d2b31e196373bb8012bee7cdf11bc70","_number":672692,"virtual_id_number":672692,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-07-26 12:08:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-07-26 08:53:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-26 08:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-07-26 10:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 08:53:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-26 08:55:29.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-07-26 12:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"863201c9a972ea17cf3283aebd3cc81252466448","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-25 10:59:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f5daec619c91d413790f5dca68ae095f367c1dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:03:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/672692/1/check/openstack-tox-pep8/edeef4d/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/92/672692/1/check/openstack-tox-py27/4449c43/ : SUCCESS in 2m 59s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario001-tempest/372e997/ : FAILURE in 58m 32s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario002-tempest/9196a87/ : SUCCESS in 1h 00m 23s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario003-tempest/5f055f9/ : SUCCESS in 58m 50s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/92/672692/1/check/packstack-multinode-scenario002-tempest/84f0b04/ : FAILURE in 1h 40m 59s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/92/672692/1/check/packstack-fedora-integration-scenario-py3-tempest/49b1d82/ : SUCCESS in 52m 59s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/92/672692/1/check/openstack-tox-py36/65f73a8/ : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"509de57e249d4f02c2a90a64877900e626069fa7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-25 13:12:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aec6fba117e9e53476e7ff301f8574b948a17c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 14:28:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/672692/1/check/openstack-tox-pep8/cd6ae25/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/92/672692/1/check/openstack-tox-py27/7e3ae55/ : SUCCESS in 3m 53s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario001-tempest/9d0fed9/ : SUCCESS in 1h 10m 21s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario002-tempest/f1b2a29/ : SUCCESS in 1h 09m 34s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/92/672692/1/check/packstack-integration-scenario003-tempest/a14eebe/ : SUCCESS in 56m 15s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/92/672692/1/check/packstack-multinode-scenario002-tempest/0f1618f/ : SUCCESS in 1h 07m 53s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/92/672692/1/check/packstack-fedora-integration-scenario-py3-tempest/f4e3dea/ : SUCCESS in 57m 11s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/92/672692/1/check/openstack-tox-py36/84e2fcf/ : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9206082bb292ce21ac7b3b6d31789ef9bcac5d1","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-26 06:46:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a26268c4c03bcc5f2452ee0e5363c7d8294fe09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 06:46:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a4d7abcd568cb7bb9a73bfbd0dff6d4b3a228c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 08:34:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/672692/1/gate/openstack-tox-pep8/7ac1430/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/92/672692/1/gate/openstack-tox-py27/f5587b5/ : SUCCESS in 3m 25s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/92/672692/1/gate/packstack-integration-scenario001-tempest/567fc27/ : SUCCESS in 1h 00m 10s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/92/672692/1/gate/packstack-integration-scenario002-tempest/08e1ee0/ : SUCCESS in 1h 06m 42s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/92/672692/1/gate/packstack-integration-scenario003-tempest/b1561b1/ : SUCCESS in 1h 11m 45s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/92/672692/1/gate/packstack-multinode-scenario002-tempest/784c4f3/ : FAILURE in 1h 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"42e29e5b03d35f4166fa908a10f1d081813eb855","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-26 08:48:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a95c7605820316a12e14e4ad14d988e0d645eed","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-26 08:53:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9f9a654bb3e98fb99ef2d739a79c72a943966ef6","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-26 08:55:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f5818eeabe8061c7818a7c84eec1419746a573f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 10:25:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/672692/2/check/openstack-tox-pep8/5189139/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/92/672692/2/check/openstack-tox-py27/e7103cf/ : SUCCESS in 3m 42s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/92/672692/2/check/packstack-integration-scenario001-tempest/0266b05/ : SUCCESS in 1h 01m 28s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/92/672692/2/check/packstack-integration-scenario002-tempest/3f5224b/ : SUCCESS in 53m 16s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/92/672692/2/check/packstack-integration-scenario003-tempest/4263339/ : SUCCESS in 1h 33m 41s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/92/672692/2/check/packstack-multinode-scenario002-tempest/a4b6a84/ : SUCCESS in 1h 13m 11s\n- packstack-fedora-integration-scenario-py3-tempest http://logs.openstack.org/92/672692/2/check/packstack-fedora-integration-scenario-py3-tempest/46b0e6e/ : SUCCESS in 52m 14s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/92/672692/2/check/openstack-tox-py36/0697eac/ : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57cf881e243508cf5446707ee7f270006e228ee9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-26 10:49:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9390b0f5e5bcf5ddad5f250b178a06e01cb49a75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 10:49:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45d268fc6c552e4fe940d0ef085a6dfe53c8899e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 12:08:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ec6fb55e2aabd1ea3967bc5dffecafc0f98a6727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 12:08:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/672692/2/gate/openstack-tox-pep8/89b6c2b/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/92/672692/2/gate/openstack-tox-py27/830242b/ : SUCCESS in 3m 36s\n- packstack-integration-scenario001-tempest http://logs.openstack.org/92/672692/2/gate/packstack-integration-scenario001-tempest/6d4d349/ : SUCCESS in 57m 35s\n- packstack-integration-scenario002-tempest http://logs.openstack.org/92/672692/2/gate/packstack-integration-scenario002-tempest/7b438f1/ : SUCCESS in 55m 01s\n- packstack-integration-scenario003-tempest http://logs.openstack.org/92/672692/2/gate/packstack-integration-scenario003-tempest/313090d/ : SUCCESS in 1h 12m 38s\n- packstack-multinode-scenario002-tempest http://logs.openstack.org/92/672692/2/gate/packstack-multinode-scenario002-tempest/12ac4a6/ : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fb64fce18d2b31e196373bb8012bee7cdf11bc70","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-13 07:44:55.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 24566dcf2f276a242b7424975a3c625ee1a93155","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c84e3b6e849a193a2a8c9c5ab7466a5bd24b5aa2","revisions":{"7242d41dd4c0dfeaf4ae950f2b89369d07760165":{"kind":"REWORK","_number":1,"created":"2019-07-25 10:59:45.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/92/672692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/672692/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/672692/1"}}},"commit":{"parents":[{"commit":"71448218c06f8452a0f7501fc263a5f46e9f6ce1","subject":"Merge \"Use upper-constraints file when running tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/71448218c06f8452a0f7501fc263a5f46e9f6ce1"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-25 10:58:45.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-25 10:59:43.000000000","tz":120},"subject":"Add dmidecode command to logs","message":"Add dmidecode command to logs\n\nWe need it to debugs issues with jobs incorrectly detecting\nvirtualization time.\n\nChange-Id: I1e53bbdcccb6402d343c98abf2696d452717d1a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7242d41dd4c0dfeaf4ae950f2b89369d07760165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7242d41dd4c0dfeaf4ae950f2b89369d07760165"}]},"branch":"refs/heads/master"},"c84e3b6e849a193a2a8c9c5ab7466a5bd24b5aa2":{"kind":"REWORK","_number":2,"created":"2019-07-26 08:48:44.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/92/672692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/672692/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/672692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/672692/2"}}},"commit":{"parents":[{"commit":"71448218c06f8452a0f7501fc263a5f46e9f6ce1","subject":"Merge \"Use upper-constraints file when running tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/71448218c06f8452a0f7501fc263a5f46e9f6ce1"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-25 10:58:45.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-26 08:47:45.000000000","tz":120},"subject":"Force qemu emulation in CI and add dmidecode command to logs","message":"Force qemu emulation in CI and add dmidecode command to logs\n\nSome upstream infra is enabling nested virtualization but it\u0027s failing\nrandomly so let\u0027s force qemu virt_type.\n\nWe need it to debugs issues with jobs incorrectly detecting\nvirtualization time.\n\nChange-Id: I1e53bbdcccb6402d343c98abf2696d452717d1a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c84e3b6e849a193a2a8c9c5ab7466a5bd24b5aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c84e3b6e849a193a2a8c9c5ab7466a5bd24b5aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
