)]}'
{"id":"x%2Fpackstack~674221","triplet_id":"x%2Fpackstack~master~I0269db6641bca335685140d2d6439bc795cd8c46","project":"x/packstack","branch":"master","topic":"bug-1838732","hashtags":[],"change_id":"I0269db6641bca335685140d2d6439bc795cd8c46","subject":"add IPv6 support for choose_my_ip function","status":"MERGED","created":"2019-08-02 08:24:16.000000000","updated":"2019-08-02 14:02:14.000000000","submitted":"2019-08-02 13:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674221-1564753601802-bfcf0b41","meta_rev_id":"85eaffd0d05b88df228bd4798f33a3500ebbd70d","_number":674221,"virtual_id_number":674221,"owner":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":2,"date":"2019-08-02 13:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-08-02 12:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-02 10:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"date":"2019-08-02 08:28:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-08-02 12:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"date":"2019-08-02 08:28:40.000000000","_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 08:28:40.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"},{"updated":"2019-08-02 08:28:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-08-02 10:14:00.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-08-02 12:22:55.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-08-02 13:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f884771162a848bcb843133119bb51d11fb72a4","author":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"date":"2019-08-02 08:24:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"236218fbca90ebce1f292f38758335172f2276fc","author":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"date":"2019-08-02 08:25:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e70d71e33da7dda5b5456b15cd0b091d7d485d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 09:47:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/21/674221/2/check/openstack-tox-pep8/37c1863/ : SUCCESS in 2m 57s\n- openstack-tox-py27 https://logs.opendev.org/21/674221/2/check/openstack-tox-py27/01a65e4/ : SUCCESS in 3m 52s\n- packstack-integration-scenario001-tempest https://logs.opendev.org/21/674221/2/check/packstack-integration-scenario001-tempest/df4805e/ : SUCCESS in 1h 08m 14s\n- packstack-integration-scenario002-tempest https://logs.opendev.org/21/674221/2/check/packstack-integration-scenario002-tempest/1974959/ : SUCCESS in 1h 02m 24s\n- packstack-integration-scenario003-tempest https://logs.opendev.org/21/674221/2/check/packstack-integration-scenario003-tempest/18713be/ : SUCCESS in 59m 10s\n- packstack-multinode-scenario002-tempest https://logs.opendev.org/21/674221/2/check/packstack-multinode-scenario002-tempest/ea31440/ : SUCCESS in 1h 08m 12s\n- packstack-fedora-integration-scenario-py3-tempest https://logs.opendev.org/21/674221/2/check/packstack-fedora-integration-scenario-py3-tempest/77bf056/ : SUCCESS in 50m 30s (non-voting)\n- openstack-tox-py36 https://logs.opendev.org/21/674221/2/check/openstack-tox-py36/e783fd9/ : SUCCESS in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc9de4db80720513fa3c30118eb8e4bbd3d01654","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-02 10:14:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8183a0ec0de23a48c25166e1408d17faba7c0aea","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-08-02 12:22:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e86c111b780d8054c54afef77c213e33bdd4663","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:23:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f11934415ad333ecb476221d4f358df75f8638e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 13:46:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/21/674221/2/gate/openstack-tox-pep8/a11846c/ : SUCCESS in 3m 32s\n- openstack-tox-py27 https://logs.opendev.org/21/674221/2/gate/openstack-tox-py27/7b4b675/ : SUCCESS in 3m 41s\n- packstack-integration-scenario001-tempest https://logs.opendev.org/21/674221/2/gate/packstack-integration-scenario001-tempest/021a384/ : SUCCESS in 1h 15m 28s\n- packstack-integration-scenario002-tempest https://logs.opendev.org/21/674221/2/gate/packstack-integration-scenario002-tempest/c099637/ : SUCCESS in 1h 10m 15s\n- packstack-integration-scenario003-tempest https://logs.opendev.org/21/674221/2/gate/packstack-integration-scenario003-tempest/59b5648/ : SUCCESS in 55m 23s\n- packstack-multinode-scenario002-tempest https://logs.opendev.org/21/674221/2/gate/packstack-multinode-scenario002-tempest/fae92f1/ : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e6fb301cb61cc7fefb68f17b91445b1a0ba4ce90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 13:46:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da53b694ff34fe0261f1cc4103e2d978727f33ba","author":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"date":"2019-08-02 14:02:14.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 894f821c1dee76909fd2697671edcdd623fa9e7b","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02bd30d9438ab2384c69c268ca564b6e7ce81182","revisions":{"5a339411a026bd813178c6447151decf8f770f02":{"kind":"REWORK","_number":1,"created":"2019-08-02 08:24:16.000000000","uploader":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"ref":"refs/changes/21/674221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/21/674221/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/21/674221/1"}}},"commit":{"parents":[{"commit":"310a9733e42b5033f0d3cef1f414d01a5b9d8918","subject":"Make enable_rabbitmq uses implied file namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/310a9733e42b5033f0d3cef1f414d01a5b9d8918"}]}],"author":{"name":"Guo Larry","email":"larry.guo@nokia-sbell.com","date":"2019-08-02 08:18:04.000000000","tz":480},"committer":{"name":"Guo Larry","email":"larry.guo@nokia-sbell.com","date":"2019-08-02 08:18:04.000000000","tz":480},"subject":"add IPv6 support for choose_my_ip function","message":"add IPv6 support for choose_my_ip function\n\nSeveral place need to use choose_my_ip function to know which IP offered\nto packstack belongs to current node. Unfortunately, IPv6 version failed.\nWith this change, it will be OK no matter which IP version given.\nFor both ovs_agent and ovn_agent, add split \":\" matching besides dot to\nsupport IPv6.\n\nChange-Id: I0269db6641bca335685140d2d6439bc795cd8c46\nSigned-off-by: Guo Larry \u003clarry.guo@nokia-sbell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5a339411a026bd813178c6447151decf8f770f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5a339411a026bd813178c6447151decf8f770f02"}]},"branch":"refs/heads/master"},"02bd30d9438ab2384c69c268ca564b6e7ce81182":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-02 08:25:55.000000000","uploader":{"_account_id":27550,"name":"GUO Larry","email":"larry.guo@nokia-sbell.com","username":"larrygu"},"ref":"refs/changes/21/674221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/21/674221/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/21/674221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/21/674221/2"}}},"commit":{"parents":[{"commit":"310a9733e42b5033f0d3cef1f414d01a5b9d8918","subject":"Make enable_rabbitmq uses implied file namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/310a9733e42b5033f0d3cef1f414d01a5b9d8918"}]}],"author":{"name":"Guo Larry","email":"larry.guo@nokia-sbell.com","date":"2019-08-02 08:18:04.000000000","tz":480},"committer":{"name":"Guo Larry","email":"larry.guo@nokia-sbell.com","date":"2019-08-02 08:24:59.000000000","tz":480},"subject":"add IPv6 support for choose_my_ip function","message":"add IPv6 support for choose_my_ip function\n\nSeveral place need to use choose_my_ip function to know which IP offered\nto packstack belongs to current node. Unfortunately, IPv6 version failed.\nWith this change, it will be OK no matter which IP version given.\nFor both ovs_agent and ovn_agent, add split \":\" matching besides dot to\nsupport IPv6.\n\nCloses-Bug: #1838732\n\nChange-Id: I0269db6641bca335685140d2d6439bc795cd8c46\nSigned-off-by: Guo Larry \u003clarry.guo@nokia-sbell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/02bd30d9438ab2384c69c268ca564b6e7ce81182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/02bd30d9438ab2384c69c268ca564b6e7ce81182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
