)]}'
{"id":"x%2Fpackstack~676123","triplet_id":"x%2Fpackstack~stable%2Fstein~I1e53bbdcccb6402d343c98abf2696d452717d1a9","project":"x/packstack","branch":"stable/stein","hashtags":[],"change_id":"I1e53bbdcccb6402d343c98abf2696d452717d1a9","subject":"Force qemu emulation in CI and add dmidecode command to logs","status":"MERGED","created":"2019-08-13 07:44:55.000000000","updated":"2019-08-13 13:14:15.000000000","submitted":"2019-08-13 13:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676123-1565702055870-674d1078","meta_rev_id":"65318cb94eafc3d866138a5d59375f4c33363e87","_number":676123,"virtual_id_number":676123,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-08-13 13:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-13 09:54:58.000000000","_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2019-08-13 08:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2019-08-13 08:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 08:51:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-08-13 13:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbe83ca1ea70b3ca39e5d0545c8b638ead075c24","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-13 07:44:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b278e0ea4474dc7429ca7ecc37569221c6885efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 08:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bf17ad875742108dc9503cfb93103a : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0162567a8b924c18942dd44694d5a737 : SUCCESS in 3m 07s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/2d5b49773f674e83b55ecafe53202a7c : SUCCESS in 44m 27s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/30e56ab812754c6e8661ad3abb00a380 : SUCCESS in 45m 39s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/45b43522753748b2957f3ae8309165fe : SUCCESS in 58m 07s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2fcd1745aec14c6895ce1bf0781a516e : SUCCESS in 47m 38s\n- packstack-fedora-integration-scenario-py3-tempest https://zuul.opendev.org/t/openstack/build/246bb3c7eafc44999486898aa89576f8 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eae4f4c112847d7b46a8726db01f2f9 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94997a862f0e46e81799e54bbd89e6f9196a4f38","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-08-13 08:51:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be0743c740ce1ffff27f74a9ee1ff8d587c8cd35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 08:51:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6f1d390ed71fb38f7af753b2202d4ef6cd47779","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 09:52:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4d9d821ac64694b79d12aeaaf96c6e : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b94177b02e2c43da8239ce5d5de5b458 : SUCCESS in 3m 18s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/dcf7778c35f44a5c9afd4a0649cf0f5d : POST_FAILURE in 3m 59s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/65f5858d1f1747618a2c7706c7a30f64 : SUCCESS in 42m 45s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/3e9a6c7ba2b04be5bc7d0ab2ae84cb79 : SUCCESS in 55m 16s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5eff7c5c4b534556ad864978c85273b0 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"64519d9e69822df69ebafc699a89f66dc77af94c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-13 09:54:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e361464e7e2780d4571086c392c0db83b22d2707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 11:05:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e890d4cdad304bc18bace50ba76cd859 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af7856b3983748d6827f7bb87efa235a : SUCCESS in 3m 30s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/dbbb9df64bc542d5a9331420cb7e4248 : SUCCESS in 44m 14s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/73b7096ce8634bf7b92ac86b9e8b618e : SUCCESS in 46m 44s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/b849c7f939204f2abf3c34dfae5dd50e : SUCCESS in 49m 24s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fc1fe52940994e4dbb7c4998695956b6 : SUCCESS in 47m 45s\n- packstack-fedora-integration-scenario-py3-tempest https://zuul.opendev.org/t/openstack/build/611e104939d847d7bd6fa52fba3ff181 : SUCCESS in 49m 23s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf25fc0efc734a1e88402489c2ca6061 : SUCCESS in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d0b020759a4ff3a633339223bac9fe757d58f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 11:05:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1ab0a23ebf203f12da35c3c4b87a0e00476061a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 13:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a26a671be0407d97fdadccfdbf6a22 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1541736fef34ea9bafda712d7d460c4 : SUCCESS in 2m 26s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/180719c904354dfda452a05382effda2 : SUCCESS in 50m 48s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1600347496d94956ba07e9cc93de0175 : SUCCESS in 58m 58s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/e39e8bc65e154f769a8252a9bbc42847 : SUCCESS in 47m 50s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cfda87b34fb84670acf09d48547ea3db : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"65318cb94eafc3d866138a5d59375f4c33363e87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 13:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24566dcf2f276a242b7424975a3c625ee1a93155","revisions":{"24566dcf2f276a242b7424975a3c625ee1a93155":{"kind":"REWORK","_number":1,"created":"2019-08-13 07:44:55.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/23/676123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/23/676123/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/23/676123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/23/676123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/23/676123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/23/676123/1"}}},"commit":{"parents":[{"commit":"c68b79943c028f2255559198058ef959059b87bd","subject":"Add OS_REGION_NAME when keystone api is \u003e\u003d 2 in rc file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c68b79943c028f2255559198058ef959059b87bd"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-25 10:58:45.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-08-13 07:44:55.000000000","tz":0},"subject":"Force qemu emulation in CI and add dmidecode command to logs","message":"Force qemu emulation in CI and add dmidecode command to logs\n\nSome upstream infra is enabling nested virtualization but it\u0027s failing\nrandomly so let\u0027s force qemu virt_type.\n\nWe need it to debugs issues with jobs incorrectly detecting\nvirtualization time.\n\nChange-Id: I1e53bbdcccb6402d343c98abf2696d452717d1a9\n(cherry picked from commit c84e3b6e849a193a2a8c9c5ab7466a5bd24b5aa2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/24566dcf2f276a242b7424975a3c625ee1a93155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/24566dcf2f276a242b7424975a3c625ee1a93155"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
