)]}'
{"id":"x%2Fpackstack~679664","triplet_id":"x%2Fpackstack~stable%2Fstein~I10f444de465437db31efd438b07ca18497580d82","project":"x/packstack","branch":"stable/stein","topic":"fedora-latest","hashtags":[],"change_id":"I10f444de465437db31efd438b07ca18497580d82","subject":"Remove Fedora 28 job","status":"MERGED","created":"2019-09-02 15:26:48.000000000","updated":"2019-09-09 13:03:24.000000000","submitted":"2019-09-09 13:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679664-1568034204484-71632aeb","meta_rev_id":"b8c3744ad0e51d7ccef5494302bb4ef7cd48910e","_number":679664,"virtual_id_number":679664,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-02 17:20:38.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-09-09 13:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2019-09-09 08:26:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-02 15:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2019-09-05 07:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-09-04 12:36:44.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2019-09-04 12:36:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 07:56:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 12:36:41.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-09-04 12:36:44.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-09-05 07:51:19.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-09-09 07:56:17.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-09-09 08:26:08.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-09 13:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9b3c01325bf763579e9afc3536b168e4c88fbe4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-02 15:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcd4512fb8abd578f75b99284c600eaa78a7ce48","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-02 15:32:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc34fa479b2d95a482ec690320edb902f75e1a8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 16:59:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62970cd150f1481781fb5a59d2c59575 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ca0992e25ab4ca19fdd6034e08c04af : SUCCESS in 4m 36s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/72718d6872f448958a0654bdb66246f5 : SUCCESS in 1h 30m 12s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/af9529b6bfba491680f884950b09704b : SUCCESS in 1h 17m 51s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/635a5394f48142e9a9e2a74b1bb2a841 : POST_FAILURE in 9m 54s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a801c3a4fbee43ebb321383528a4f1e3 : POST_FAILURE in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66fdef7e42e9427cabcb84f16b8ac07d : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"180f2a1bb1ce50a936e07a57a4cfd7f626796904","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-02 17:20:05.000000000","message":"Patch Set 1:\n\nany ideas why the post playbooks fail?","accounts_in_message":[],"_revision_number":1},{"id":"722cace8286468b3d2e3377507811046d1cf2f4b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-02 17:20:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4e729783cef31b29191e5c0c0ebe8527bd53ebc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-02 17:20:45.000000000","message":"Topic changed from stein/backport/679115 to fedora-latest","accounts_in_message":[],"_revision_number":1},{"id":"4c3f8857340ee3f00eb0ef56a185061bc699f939","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 17:38:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a489a6de8043a0a4405ea35cc7770f : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5319611e8b8646fbbf0987728e1357bc : SUCCESS in 3m 51s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/5677c411a6c140b7b2e4de4b0ab4de75 : POST_FAILURE in 7m 32s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5ab1aba2e542492cbc9144a230864b46 : POST_FAILURE in 10m 50s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/12ae0c19fa6642cdb9c8a1bca60dd6c3 : POST_FAILURE in 7m 35s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7d730348ed7f4bc6b277f0b1e18e4275 : POST_FAILURE in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18a323eff8c54b2daf9caac3b2243fba : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2afff47742a3620b636da7353f43402dd43ae36b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-03 10:43:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d5d5d15e5d8f835b7e4e0a3e90c6d5142e913d6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:27:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33768c3cb3fa4a17871edecc735bd526 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfd1709bcf2f45fcb3d7393fcd18445f : SUCCESS in 3m 36s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8cd90ae854d8416e8fe47d871ac236dc : SUCCESS in 1h 20m 26s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1bcc2cfa3d9b4a42bc6fdeb62f829d8a : POST_FAILURE in 14m 52s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a66fd5856c27481dbd892ca9558a43c6 : SUCCESS in 1h 19m 25s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/13c409d2cda144e9bd7fa1e2c84f67d2 : FAILURE in 1h 13m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26487bbd986f4a3c9777b11242a03071 : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6fc404b9629b8b8c68f5b07049d64a23805bac3","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-04 09:39:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43f583ede2cecc5e01ca9c7f9da834303d8e74d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e3b7e746814671baed8ee2f62fd976 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8190c4db29b4b53ba5bb3dcb80896a4 : SUCCESS in 3m 58s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/665fe09dbb044529944afc5b25a71fc6 : SUCCESS in 1h 08m 52s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/962b5b486da1419385da6850a9f7f9d5 : SUCCESS in 1h 11m 06s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/2f70043b200949198906a12eedd1d73e : SUCCESS in 1h 12m 35s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/58700f64d4d4489798f675e05e00537a : SUCCESS in 1h 24m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2394400582fd4d49ac29cc62ba04164a : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aea66fc74fbdedb133fee602e518ef038646999","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-05 07:51:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c444a38fffbff6a0e170f05f6c1e25d82592bed6","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-09 07:56:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1826d30a1a08bd9b2210217b8478b540691ea6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 07:56:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af48a488802bbdcf6a5978daa94d75604216709b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-09 08:26:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b3fd31005ad52b3d7234543c51ed82bf07b85542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 09:46:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54764244259439b9f1b00a7d4d9acc7 : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a69c655c3c784068a56635f5641fc34f : SUCCESS in 4m 05s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/ae2b52d2516b4c829a703ad60f4b9909 : SUCCESS in 1h 09m 30s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/052f64a24d4c41dd87ab22afb0ceab0f : SUCCESS in 1h 25m 59s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/61751674379b4ee1925cba610d4e350d : SUCCESS in 1h 26m 39s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1fcbe102502d4ce094fb98ddb4f361c9 : POST_FAILURE in 1h 44m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b959e54230435d24734659ce5e242d6c863dd645","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-09 09:50:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6eca048b39df1281b94703a4f3153a83d272aab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d7a672353e4f56a230dd01dd6fa61d : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7d8ca5a3c3242ffac96e3372f85a55f : SUCCESS in 3m 22s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8b65e602b7d3493984b90a52b257d6af : SUCCESS in 1h 07m 32s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e202430830be41ed9eb84fb755610ddc : SUCCESS in 1h 17m 07s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/1d8ce749ad884f9da1bd45dc7e8296cd : SUCCESS in 1h 15m 37s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5dfd85ca108e4a158111c9e2a8772a8a : SUCCESS in 1h 28m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc59f1703894cfdb2a0ed852db16288 : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e844d478e20f99ef83627b20babc3ca9e09b7f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:26:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6bd4f2a56928d6f4866f7cf7c2528e2bb7b9ee2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b196e5a6c5947bd9f0429a29987347d : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3b79eb252144f6da0c03ed273efd010 : SUCCESS in 3m 02s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/5c3424048c564e8b809005bb648d7395 : SUCCESS in 1h 10m 34s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1b98e249bbb048c08eee050fca499312 : SUCCESS in 1h 11m 05s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/1525628c0c9f4ee1a316db88c0d70006 : SUCCESS in 1h 28m 41s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/27fcf21c483e4145bb8a7d7a8d7e71e3 : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b8c3744ad0e51d7ccef5494302bb4ef7cd48910e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa8e8fdab1b2cd246ad7d18d3f919856672a416d","revisions":{"fa8e8fdab1b2cd246ad7d18d3f919856672a416d":{"kind":"REWORK","_number":1,"created":"2019-09-02 15:26:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/64/679664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/64/679664/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/64/679664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/64/679664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/64/679664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/64/679664/1"}}},"commit":{"parents":[{"commit":"9bf6c112021e246da6694f72f86c646140d5c45b","subject":"Merge \"change memcached IPv6 bind host to IPv6 format\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9bf6c112021e246da6694f72f86c646140d5c45b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-28 18:59:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-02 15:26:28.000000000","tz":120},"subject":"Remove Fedora 28 job","message":"Remove Fedora 28 job\n\nFedora 28 is end-of-live and gets removed from OpenDev testing,\nremove it completely here. Packstack plans to use CentOS 8 instead of\nFedora for Python 3 based distros.\n\nThis will be backported to all open branches.\n\nChange-Id: I10f444de465437db31efd438b07ca18497580d82\n(cherry picked from commit 28c795eccf21a7f849e0c4cae713494e56458f5f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa8e8fdab1b2cd246ad7d18d3f919856672a416d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa8e8fdab1b2cd246ad7d18d3f919856672a416d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
