)]}'
{"id":"x%2Fpackstack~681378","triplet_id":"x%2Fpackstack~master~I8f1f3f39b7f4648ac2cb6ed8caceb48d9dba0ad0","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I8f1f3f39b7f4648ac2cb6ed8caceb48d9dba0ad0","subject":"Remove usage of deprecated cinder parameters","status":"MERGED","created":"2019-09-11 05:32:20.000000000","updated":"2019-09-11 15:32:19.000000000","submitted":"2019-09-11 15:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681378-1568215939687-9806e538","meta_rev_id":"f1eeb1d6b0fee25e20604fbfb8c6270446b8bf76","_number":681378,"virtual_id_number":681378,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-09-11 15:32:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-11 08:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"date":"2019-09-11 08:01:25.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"date":"2019-09-11 08:01:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-11 13:46:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 08:01:22.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-09-11 08:01:25.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-09-11 13:46:28.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-09-11 15:32:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"855b54bf22624a3f106dc9119d5f69216c2672cb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-11 05:32:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f55fc89d24548387905024566429ea3d04e1a58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 07:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5645c4189b94cf3bd99bf4666f99470 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a01467f6c7e4b44ace7a85ba9642f3c : SUCCESS in 3m 14s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/fdd59418140549c783cda880fc739d84 : SUCCESS in 1h 10m 33s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2ec644ca2af34509ad08c733e311a7ba : SUCCESS in 1h 10m 55s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/eeb1522ada42400298c1357014d02a63 : SUCCESS in 1h 15m 49s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ed5bfa5143fd4f488e544f87f916c992 : SUCCESS in 1h 33m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3f7bb44703c4e80988fdabd08855f70 : SUCCESS in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ea2d46b6fc1fba0da2eb58938a976dbdff682ff","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-11 08:01:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d26510743d2d6ed9a582474e9c64de1db63f4a0","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-11 13:46:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"092b534d3752b484b616727069f84466a5b31e64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 13:46:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"295f8d6c74924c24676f56043b65c3bad28f061e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:32:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650e040914124518a610118f1db198ac : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f30c88fe09a4bc487cc09ae0a2ae5aa : SUCCESS in 3m 20s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/18d2fc2fd2394f3f90c9139971b68b0f : SUCCESS in 1h 13m 44s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0e584d4d747945e398d3a9a7ec758e9f : SUCCESS in 1h 25m 01s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/833961a7bb404a348c6604b40292ab16 : SUCCESS in 1h 36m 26s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/71cd70c361ab4cca9df514a020679b42 : SUCCESS in 1h 39m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f1eeb1d6b0fee25e20604fbfb8c6270446b8bf76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:32:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"000b695f9e038619d09433fb640233e26344f3fc","revisions":{"000b695f9e038619d09433fb640233e26344f3fc":{"kind":"REWORK","_number":1,"created":"2019-09-11 05:32:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/78/681378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/78/681378/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/78/681378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/78/681378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/78/681378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/78/681378/1"}}},"commit":{"parents":[{"commit":"ac2c9f021f6d5261ddaa463e9dd5c2114ade6c92","subject":"Merge \"Fix setup.py for Pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ac2c9f021f6d5261ddaa463e9dd5c2114ade6c92"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-09-11 05:30:20.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-09-11 05:30:22.000000000","tz":330},"subject":"Remove usage of deprecated cinder parameters","message":"Remove usage of deprecated cinder parameters\n\nhttps://review.opendev.org/#/c/678965 removed\nsome deprecated parameters for puppet-cinder,\nthis patch cleans there usage in packstack, also\nreplace deprecated cinder::type usage with\ncinder_type resource.\n\nChange-Id: I8f1f3f39b7f4648ac2cb6ed8caceb48d9dba0ad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/000b695f9e038619d09433fb640233e26344f3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/000b695f9e038619d09433fb640233e26344f3fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
