)]}'
{"id":"x%2Fpackstack~681486","triplet_id":"x%2Fpackstack~stable%2Fstein~I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af","project":"x/packstack","branch":"stable/stein","hashtags":[],"change_id":"I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af","subject":"Enable replication to redhat-openstack/packstack with upload-git-mirror","status":"MERGED","created":"2019-09-11 14:00:29.000000000","updated":"2019-09-17 14:49:57.000000000","submitted":"2019-09-17 14:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681486-1568731797772-795c03e7","meta_rev_id":"a18f8fdb77469bcd7453841de9dbb68567e1015f","_number":681486,"virtual_id_number":681486,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-09-17 14:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2019-09-11 16:22:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-09-17 08:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"date":"2019-09-11 14:00:29.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-09-17 08:40:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 14:00:29.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-09-11 16:22:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-17 08:40:09.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-09-17 14:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e086d0f81a8311adc69eceb1641bd99857f72ff8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-11 14:00:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf87366a5c0f9715d19d86b237b4a0f89f628eeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:52:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f244dfff43e743c69f2b4d037890db68 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e79b29795ed43da9c581c1c6dc3b94b : SUCCESS in 3m 17s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/9ad31e63f9384167a154ff9d969efab2 : SUCCESS in 1h 14m 36s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c3f8cbfbdc52430d861ea9ace7e79027 : SUCCESS in 1h 29m 40s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/7d03ce04a5fa4521bad1d7218f237a8d : SUCCESS in 1h 16m 09s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/552f6d04656049b5bba4a80d4fd33c01 : POST_FAILURE in 1h 46m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51e0dff3940e40d794ba47c416dc074f : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9d7d57829937b20453badd4ce9ca8141de82033","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-11 16:22:13.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0767318cd5e617dc25599806f6e0f34a2ce139d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:34:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e46645b3b14746b4b403508544966117 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e488e0b72d5f4ab08cbb84a9434c3e95 : SUCCESS in 3m 43s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8452b6f3920f4c428b9b2e25f669a2ac : SUCCESS in 1h 10m 36s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5521cb65b52944bea99bd2bbc2d647e4 : SUCCESS in 1h 06m 47s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/85db59fbba8d4fb3bba23b598bf81e60 : SUCCESS in 1h 31m 19s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/19acc678bcec4ffebbaf658a4b4bbf5f : SUCCESS in 1h 30m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/566e260b1dc9424eb7995c914ef0d2e8 : SUCCESS in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"380507e2e1494004542cdc77910881278d8294f7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-09-17 08:40:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71b15912a94cd969f954f80fa2eb828548eac667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaf80ba519febd531d8a17496c7f81958cfe2d78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:34:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63db6651eb2a46fab5689155d675caa9 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/256eab98ab9b4878a61a01cfa53ee1c8 : RETRY_LIMIT in 3m 33s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/ebb9437f6d3e4b5794a00b608cbe40e9 : SUCCESS in 1h 23m 29s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b32037d8f9e54fc3b575c59e3926fa02 : SUCCESS in 1h 08m 01s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/b80aacf4e48346e0bf0d8a1595142ef7 : SUCCESS in 1h 34m 39s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d827d3c9a6d2471d8b1789f6e2d51ed6 : POST_FAILURE in 1h 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"40aa846ef7d0246d147fe7bb20d546cac7e3bf37","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-17 10:52:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb2295542981f48dda4a31c69bf8594fb6a8ecf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e23efcdc8f4335b477c50782aa37f4 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bde4d30c9ef440bbada237352cc8b248 : SUCCESS in 3m 46s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/49b0ce8817b94aceaaf7763d07f2e329 : SUCCESS in 1h 17m 40s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fd668250806b4ba5af9be31f3943a978 : SUCCESS in 1h 26m 29s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/045b8b37869e4293bda1eeafd361bcfe : SUCCESS in 1h 18m 02s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7db2cd08e9624b71b2b92f1f7a1bfaf0 : SUCCESS in 1h 41m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a90afc85e746d1afba4971693f29b9 : SUCCESS in 3m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fbb67f17c411e5f394854b5a00485d1b71e242d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:39:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8fe2f3293915ca313fb725e3abb8174deb2232c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:49:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab327093f9fc428abf4028a09a5a5624 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9344dce7d1604168bbd7bd95bc4ab833 : SUCCESS in 2m 55s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/b8cb8c147f404a3f8ae391d2879646cf : SUCCESS in 1h 11m 48s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fa35179a778b4617bc39635d84836080 : SUCCESS in 1h 37m 41s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/dc12aea9d29a4b19894f3422ba1c68ee : SUCCESS in 1h 15m 51s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/315228ac2b9b4b4990fa5af045a1a982 : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a18f8fdb77469bcd7453841de9dbb68567e1015f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:49:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd85c4eebc2acb5344e9b02304aaad5ef031c552","revisions":{"cd85c4eebc2acb5344e9b02304aaad5ef031c552":{"kind":"REWORK","_number":1,"created":"2019-09-11 14:00:29.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/86/681486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/86/681486/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/86/681486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/86/681486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/86/681486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/86/681486/1"}}},"commit":{"parents":[{"commit":"fa8e8fdab1b2cd246ad7d18d3f919856672a416d","subject":"Remove Fedora 28 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa8e8fdab1b2cd246ad7d18d3f919856672a416d"}]}],"author":{"name":"David Moreau Simard","email":"dmsimard@redhat.com","date":"2019-05-07 14:42:36.000000000","tz":-240},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-11 14:00:21.000000000","tz":120},"subject":"Enable replication to redhat-openstack/packstack with upload-git-mirror","message":"Enable replication to redhat-openstack/packstack with upload-git-mirror\n\nThe replication to github is no longer automatically managed by opendev.\n\nChange-Id: I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af\n(cherry picked from commit 51fb71534a46b3d3c06f08241806a8d3d2a4cd93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cd85c4eebc2acb5344e9b02304aaad5ef031c552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cd85c4eebc2acb5344e9b02304aaad5ef031c552"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
