)]}'
{"id":"x%2Fpackstack~681488","triplet_id":"x%2Fpackstack~stable%2Frocky~I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af","project":"x/packstack","branch":"stable/rocky","hashtags":[],"change_id":"I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af","subject":"Enable replication to redhat-openstack/packstack with upload-git-mirror","status":"MERGED","created":"2019-09-11 14:01:54.000000000","updated":"2019-09-17 12:14:50.000000000","submitted":"2019-09-17 12:14:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681488-1568722490259-b01d7e93","meta_rev_id":"a77c4be0bc4937b025ca66e30c06194b513b2dd1","_number":681488,"virtual_id_number":681488,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-09-17 12:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 16:03:12.000000000","_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-09-17 08:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"date":"2019-09-11 14:01:54.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"date":"2019-09-11 16:03:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-09-17 08:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 14:01:54.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-09-11 16:03:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-17 08:40:29.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-09-17 12:14:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db78fe675a89f385561be77f4eb60db341b75b55","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-11 14:01:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"316094d264c3644f15429d89d185214e86c7465c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:43:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459eb4bbce4240df958fb1715d02a87c : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02d09dc86b354b01925a76d5cc1f2207 : SUCCESS in 2m 49s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/2be665e45c4a4483808b43754a76c270 : SUCCESS in 1h 24m 19s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2b5e930128674cf5a8f5e0cd6519e019 : SUCCESS in 1h 11m 25s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/95ce1f341004419bbe7e6195e49ef34e : SUCCESS in 1h 14m 49s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/836ce1e344a848458d5e84315bf2d221 : POST_FAILURE in 1h 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"da7888cf0957c3661371cb2ee8f3cc68476b33e5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-11 16:03:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85761c4d2ad16748d6b013841adab09f0553e8f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 17:57:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4bd205f2b14436996ff0202838ae35 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84964e8c3b194361a3a7cdbc2dd7ce5f : SUCCESS in 6m 07s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/71f1c7af544f4d5290590657894fbbd7 : SUCCESS in 1h 11m 38s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ed1b8b20413f4d569c1f671996ee4c88 : SUCCESS in 1h 09m 27s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ce2db36121764786b22668ff1b2e96e0 : SUCCESS in 1h 22m 51s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a33aef772e6e4a38a629e6a0fcce7e14 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8fe78cc2c5b0ed6b5482200c5bea7239425008ed","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-09-17 08:40:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c366916efa20ed8253dea9bdd88555f3efd13336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:40:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9259c644f81c1d2b533c7ca2280023df0ae2f49c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:14:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d959a567c3c4d2cab7a31fa88fd1b58 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be5c55732ed944aca25a784a37879af5 : SUCCESS in 3m 54s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/b4492245e1e34e83b458bb9b4cb8dd52 : SUCCESS in 1h 17m 31s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/02c3e71f4f644a5386354fca16778448 : SUCCESS in 1h 10m 02s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/3466b5c53c45441f844c8a616c05e092 : SUCCESS in 1h 18m 50s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0dd33b981d41493c8bebf5b940acda50 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a77c4be0bc4937b025ca66e30c06194b513b2dd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:14:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28395d660f763594b12931c6894621f2366f0009","revisions":{"28395d660f763594b12931c6894621f2366f0009":{"kind":"REWORK","_number":1,"created":"2019-09-11 14:01:54.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/681488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/88/681488/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/88/681488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/88/681488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/88/681488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/88/681488/1"}}},"commit":{"parents":[{"commit":"09bf673a3ff82d86747f854c488e7f96837b8f76","subject":"Fix URL to get the list of projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/09bf673a3ff82d86747f854c488e7f96837b8f76"}]}],"author":{"name":"David Moreau Simard","email":"dmsimard@redhat.com","date":"2019-05-07 14:42:36.000000000","tz":-240},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-11 14:01:47.000000000","tz":120},"subject":"Enable replication to redhat-openstack/packstack with upload-git-mirror","message":"Enable replication to redhat-openstack/packstack with upload-git-mirror\n\nThe replication to github is no longer automatically managed by opendev.\n\nChange-Id: I76eb105f6cd2624b6c9edfc0dcedd2f2c56aa5af\n(cherry picked from commit 51fb71534a46b3d3c06f08241806a8d3d2a4cd93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/28395d660f763594b12931c6894621f2366f0009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/28395d660f763594b12931c6894621f2366f0009"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
