)]}'
{"id":"x%2Fpackstack~687268","triplet_id":"x%2Fpackstack~master~Ib28328c388cc4b30eeaf111a6a1abd1f56e5f575","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Ib28328c388cc4b30eeaf111a6a1abd1f56e5f575","subject":"Open Ussuri","status":"MERGED","created":"2019-10-08 11:08:15.000000000","updated":"2019-10-08 17:11:42.000000000","submitted":"2019-10-08 17:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687268-1570554702370-6aa585c6","meta_rev_id":"724039a5a4c924c78e06b7092d3611ed577a4eed","_number":687268,"virtual_id_number":687268,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 17:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":2,"date":"2019-10-08 14:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-10-08 12:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"date":"2019-10-08 11:15:38.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2019-10-08 14:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-10-08 12:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 11:15:38.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-10-08 12:56:57.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-10-08 14:11:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-10-08 17:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"766a6fd8426d3d8936d3e645bbc0fad537a7527c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-08 11:08:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32df503bb7d887c285d864e2ddf82fa6f39f6680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:42:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c493bc84b3d415586643144b9b22d9a : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/827825c5009941c2bccc23051257dae6 : SUCCESS in 2m 52s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/826f493142c248aabd3339d8c413cc64 : SUCCESS in 1h 11m 38s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9f58eac430f54e4b9f4764aac4567ad4 : SUCCESS in 59m 11s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/35b9e409548e4d7aa0090cf2ddba582a : SUCCESS in 59m 06s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a3f43667e17f4ddb987ac04fb9f4dd3b : SUCCESS in 1h 29m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fffc90716601464d8fc6754f498ce5a8 : SUCCESS in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c889dabdf59b690c56df5bd7334609249b615c4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-10-08 12:56:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cb17d3e1ab0d2d99140055069eab58a1bda5a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:57:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cef99ad81a9225371e8fabecbf3858dea828487f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:07:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d489e85dba4a7f9d318d8ecf8df76c : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d87dde0b8cbd4e738622be29863039f7 : SUCCESS in 3m 01s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/31e6152407f34f4db9b9d41d7f43ccbc : SUCCESS in 54m 17s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/63938fb383a04b6d9243d71fd4d99165 : FAILURE in 1h 07m 51s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/d3bfef03434f4f2a9401601cc595f672 : SUCCESS in 1h 09m 51s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/83053a7feeb549ff9dde34b4ae952b87 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":1},{"id":"41f10355f7d3d5367f7135749643627c5e48684f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-08 14:11:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7bb51b91275b32b44785e86236239c04d38cdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c876f0c318374f2c83f6e090aad340f1 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8da222f7971946b4bc3fcbbd46a59160 : SUCCESS in 5m 36s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/5a1dc4e7734c49cea284e75a0b5853f9 : SUCCESS in 1h 17m 33s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/80c2510f442944059cb1ab0a76924e42 : SUCCESS in 1h 10m 00s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/5b61affbfec64b319f10b160e5cfd782 : SUCCESS in 1h 17m 42s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/78878dfedb8f4680b8f0215c89f60035 : SUCCESS in 1h 20m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/354fd64a1987434bbb6357f5f6bb141d : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"314ea1f61a04ce18b89ad4291b5a9349c620bdcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23c9e608330871952219cf9f200eeab6f998062a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 17:11:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"724039a5a4c924c78e06b7092d3611ed577a4eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 17:11:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e520fa8d17a24526ae62cc7a9b5df016 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7742df032a3d4326bf628fb8504a7143 : SUCCESS in 3m 57s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/4910e6c013774b79afed246011c0777e : SUCCESS in 1h 11m 26s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9b04767e2f88497ca29573298a2aa63b : SUCCESS in 1h 26m 48s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/bb728b5f00664a3789e29b637b038218 : SUCCESS in 1h 03m 27s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/296e78945a524037b87f300444d01232 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4c7a5eedb7145fc010f6fadc76a8db6cd527261","revisions":{"b4c7a5eedb7145fc010f6fadc76a8db6cd527261":{"kind":"REWORK","_number":1,"created":"2019-10-08 11:08:15.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/68/687268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/68/687268/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/68/687268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/68/687268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/68/687268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/68/687268/1"}}},"commit":{"parents":[{"commit":"000b695f9e038619d09433fb640233e26344f3fc","subject":"Remove usage of deprecated cinder parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/000b695f9e038619d09433fb640233e26344f3fc"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-08 11:06:59.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-08 11:08:03.000000000","tz":120},"subject":"Open Ussuri","message":"Open Ussuri\n\nThis commit will bump the major version reported by pbr, so Ussuri\ncommits will always have a higher version than Train ones.\n\nChange-Id: Ib28328c388cc4b30eeaf111a6a1abd1f56e5f575\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b4c7a5eedb7145fc010f6fadc76a8db6cd527261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b4c7a5eedb7145fc010f6fadc76a8db6cd527261"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
