)]}'
{"id":"x%2Fpackstack~691258","triplet_id":"x%2Fpackstack~stable%2Fqueens~Ib9b4194c88cd9065babb81c788ed52b2ee5b8d2f","project":"x/packstack","branch":"stable/queens","topic":"cached-cirros","hashtags":[],"change_id":"Ib9b4194c88cd9065babb81c788ed52b2ee5b8d2f","subject":"Use cirros image from cached for demo provisioning","status":"MERGED","created":"2019-10-25 08:23:55.000000000","updated":"2019-10-25 12:20:39.000000000","submitted":"2019-10-25 12:20:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691258-1572006039427-b8c1e99e","meta_rev_id":"eec0857de4b2ac8c341e3e65a87e9006980b03fa","_number":691258,"virtual_id_number":691258,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 12:20:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2019-10-25 10:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2019-10-25 10:48:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 10:48:00.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-10-25 12:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"765a6cccd29e0028be709fc1e5faa25490a0307e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-25 08:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8694ed65e4b98d0a36b75e6a3534757ea663f628","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-25 08:24:34.000000000","message":"Topic set to cached-cirros","accounts_in_message":[],"_revision_number":1},{"id":"1f22ceddc1fe4626d461edb992e7b93a3382d41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/494287e9c6644c00a54c4850f0891a8b : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fdc2de6d7034fae9cf091deb01175c7 : SUCCESS in 4m 16s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/e544f9bddd0f4148a93b384808f7ac48 : SUCCESS in 1h 01m 37s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b469b8a7b2bf4d388fc88bb5d6ec1bce : SUCCESS in 1h 04m 29s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/68d055ddeff24d4a8db8778b7b6cd58d : SUCCESS in 1h 29m 46s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d1109fcb0f024b04b2e55efb38cc62ff : SUCCESS in 1h 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ce3c30434ff393e1a8bf7ea8a1e3f4049ce328aa","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-10-25 10:48:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62447b249ea97d70198c01f0e096fe09fa3d3f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:48:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61ba94450c05996250b7490908fcfac08ff6786f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eec0857de4b2ac8c341e3e65a87e9006980b03fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:20:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f900199e9832450eadc98d44f431d7a4 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30a07b149e35454da50546113155d28f : SUCCESS in 3m 12s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/161507be05bc42288ef3e1a22cac7ec7 : SUCCESS in 1h 18m 24s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/34682bce61df4b70b21712520124a48a : SUCCESS in 1h 18m 27s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/5ccf848e13ea48f583c009875e79ecef : SUCCESS in 1h 03m 02s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bdf739cbdb514c52b5ae0b7bc4d7a42e : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98f48637cb2044489a46975fadc0808df8e339be","revisions":{"98f48637cb2044489a46975fadc0808df8e339be":{"kind":"REWORK","_number":1,"created":"2019-10-25 08:23:55.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/58/691258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/58/691258/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/58/691258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/58/691258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/58/691258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/58/691258/1"}}},"commit":{"parents":[{"commit":"db5cb2d203dfa8de51c60619eb77eed2e28b33e3","subject":"Fix setup.py for Pypi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/db5cb2d203dfa8de51c60619eb77eed2e28b33e3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-10-22 09:24:04.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-10-25 08:22:59.000000000","tz":330},"subject":"Use cirros image from cached for demo provisioning","message":"Use cirros image from cached for demo provisioning\n\nCurrently, we are using cached image for UEC (tempest) but not for demo\nprovisioning.\n\nChange-Id: I637edcdcc129dc8ed86eb4f3fdd4e216dd3f48a7\n(cherry picked from commit f9cc0729aebb6bd4d30f40f1bd7060cc8c36de25)\n\nUse https for cirros downloads\n\nhttp is being much slower and anyway using https is a good practice.\n\nChange-Id: Ib9b4194c88cd9065babb81c788ed52b2ee5b8d2f\n(cherry picked from commit 97e5a40dbbcaf616bba4dd85d56216777aa77faf)\n(cherry picked from commit 21e85e1a8c946610f82a847aa0ec6ce2f42b38c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/98f48637cb2044489a46975fadc0808df8e339be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/98f48637cb2044489a46975fadc0808df8e339be"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
