)]}'
{"id":"x%2Fpackstack~698253","triplet_id":"x%2Fpackstack~master~I22048b580fef254090c77f8ded58513b44e93c36","project":"x/packstack","branch":"master","topic":"centos8","hashtags":[],"change_id":"I22048b580fef254090c77f8ded58513b44e93c36","subject":"Use rabbitmq parameters for tcp_listen_options","status":"MERGED","created":"2019-12-10 14:11:31.000000000","updated":"2019-12-17 08:35:01.000000000","submitted":"2019-12-10 17:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698253-1575998303265-7edb6769","meta_rev_id":"29508bb9309070fa837f8bf51f8d42b9b982926f","_number":698253,"virtual_id_number":698253,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 17:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2019-12-10 16:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-10 14:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-12-10 16:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 14:29:11.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-12-10 16:18:47.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-10 17:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2bf91244da499958fa731c8a497fc8bea2ba473","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-10 14:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c22ac19673e4644ae7242ccddcae9910dc1a533","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-10 14:29:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86f82ccf08327382df85b1e26c267dbd45bdfa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 15:18:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dac11571e224f4087ede87cce6c0afd : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acf601550c784ef593dfcebd09c844ef : SUCCESS in 4m 44s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/e9d5610c829c4394b6548671627c2148 : SUCCESS in 46m 48s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/de642ed8edac40389470915553ae62e9 : SUCCESS in 51m 20s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/6d5d5eb252b14f5baad873d8905811d2 : SUCCESS in 1h 00m 47s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f45143af13bd4991ab63da6f24ad3023 : SUCCESS in 45m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eae367fc3e44be4ac2ab71dc324bf6f : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"644f9e747a68da64c59883dbd34be0ada869642b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-10 16:18:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09ec5bcd259963a69aa2518f0bcb2b120ba37ffe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:18:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b1d9a0a84650b0d5a14bdb0011a5c994a29499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:18:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6a706dc0634a8996d3f0a0111e59d7 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfdf714fcc3a4b84a19a0f3d2882a533 : SUCCESS in 5m 59s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/c728fc8e278f424a9f5f0e102170b012 : SUCCESS in 40m 04s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9a31d51a86d24ee8a88415410f1430dd : SUCCESS in 42m 08s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/56dd95953a404dbda6d673a9750ad011 : SUCCESS in 51m 52s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/44bad8d8b5d6445887857c702dcbaca6 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a65f777790c858357eba11ac3cf0b8f0f88cc6db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec1044f61fc9995698775fe5255c717562a96ae2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-17 08:35:01.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit bfd32144835e1b43a760d9a4a498c4f21eae49db","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96bd0f7ca64c95d96d64b55b5f6c443e4d24343e","revisions":{"96bd0f7ca64c95d96d64b55b5f6c443e4d24343e":{"kind":"REWORK","_number":1,"created":"2019-12-10 14:11:31.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/53/698253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/698253/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/698253/1"}}},"commit":{"parents":[{"commit":"4a1fd433e8482f58edd741ccfc70ec9fa473806b","subject":"Set unixsocket to undef in redis class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4a1fd433e8482f58edd741ccfc70ec9fa473806b"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-10 14:05:27.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-10 14:11:29.000000000","tz":60},"subject":"Use rabbitmq parameters for tcp_listen_options","message":"Use rabbitmq parameters for tcp_listen_options\n\nCurrently we are passing tcp_listen_options as part of config_variables.\nHowever, puppet-rabbitmq creates this directive from specific options [1].\nThis is leading to configuring it twice with different options. This is\nnot valid for rabbitmq 3.8 which does not start if it detects it\nconfigured twice.\n\n[1] https://github.com/voxpupuli/puppet-rabbitmq/blob/master/templates/rabbitmq.config.erb#L34-L55\n\nChange-Id: I22048b580fef254090c77f8ded58513b44e93c36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/96bd0f7ca64c95d96d64b55b5f6c443e4d24343e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/96bd0f7ca64c95d96d64b55b5f6c443e4d24343e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
