)]}'
{"id":"x%2Fpackstack~698453","triplet_id":"x%2Fpackstack~master~Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","subject":"Ensure the Aodh DB is synced","status":"MERGED","created":"2019-12-11 12:17:06.000000000","updated":"2019-12-11 17:04:50.000000000","submitted":"2019-12-11 16:53:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698453-1576083216868-a6c9ab22","meta_rev_id":"4994f0d6119cbab344ad16cee9e25260c2395152","_number":698453,"virtual_id_number":698453,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 16:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 12:21:21.000000000","_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2019-12-11 13:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-12-11 13:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},{"value":2,"date":"2019-12-11 15:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},{"value":1,"date":"2019-12-11 15:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 13:44:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-11 13:45:20.000000000","updated_by":{"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},"reviewer":{"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},"state":"REVIEWER"},{"updated":"2019-12-11 15:43:37.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-12-11 16:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a07cf64ac62edf1283e19c20fd8d8a8aa795c49d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 12:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55a262f263c63e8099d233e35bdba3293d13c9d2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 12:21:21.000000000","message":"Patch Set 1:\n\nLooking at the puppet-aodh code, it looks like we\u0027ve had this bug for a very long time :?. Once it\u0027s merged, let\u0027s backport this patch to all supported stable branches.","accounts_in_message":[],"_revision_number":1},{"id":"b36e4c36c5b6c1e060ae6867fc820493b2e3c87f","author":{"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},"date":"2019-12-11 12:38:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87ea0977cd5a6a8f49b51d0d4f75c198a09ea0b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 13:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff91162f1ea45c1abf3ec57be1bd18b : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8355d8114d342a6be613b4d39b6f073 : SUCCESS in 3m 03s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/bf444d90acb64682b06c1044f57f8025 : SUCCESS in 41m 50s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8372425b7833466b9816faf1f9b82304 : SUCCESS in 42m 45s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/8573e143777d49cdb96a98e4e37270f5 : SUCCESS in 43m 29s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/62a589323a2a44a998a48fd82895e4ac : SUCCESS in 46m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3f2c6bb9194f20acd77433189474f2 : SUCCESS in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db91e632adecdb8a2bf0a49fb4233b4ab26b965d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-11 13:44:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fb578df631829af45485c85ebcdf365001930fc","author":{"_account_id":17927,"name":"Manik Sidana","email":"manik@voereir.com","username":"maniksidana019"},"date":"2019-12-11 13:45:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b3c02bea4436d5392b7e481c51cd2ca4d41e498b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-11 15:43:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c26c1c367b15d22cbb265d975aea29952d973ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:43:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ea85c864b6d6c695c44f69a73a8d8d5d1554501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:53:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220f19bc62184056bceb30e922ad4532 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f15ad203d158435d8c9fccbf9695a404 : SUCCESS in 3m 30s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/33446bdc56184d6bbbd921282ac10d76 : SUCCESS in 39m 25s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5ef0b2140bc84fddb3d1742954a48984 : SUCCESS in 54m 43s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a3253d8b6cfa4405a28fbe5a40a6d49f : SUCCESS in 51m 41s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b40410f4098a45d58253ab24503f0c0b : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d52273dbe37ccce481f8f953fc034dfac205a4c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:53:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"476a5cbb7dd340eaf5ded6483c655931f3531e76","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 17:04:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 4f7fe32b2fc44e956898426abe44b24da543abe1","accounts_in_message":[],"_revision_number":1},{"id":"dfeddb2bd7ae2ce7f2c96c63f7c8a07df2b0a75d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 17:04:41.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 003748246f6090a79d1432727ba0817e8f48130e","accounts_in_message":[],"_revision_number":1},{"id":"4994f0d6119cbab344ad16cee9e25260c2395152","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 17:04:50.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit c024700d60b30f0a858effd16ab625fd4a269b6c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f64071f5a17dd3d66033edd5c3786d67459c293","revisions":{"1f64071f5a17dd3d66033edd5c3786d67459c293":{"kind":"REWORK","_number":1,"created":"2019-12-11 12:17:06.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/53/698453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/698453/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/698453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/698453/1"}}},"commit":{"parents":[{"commit":"96bd0f7ca64c95d96d64b55b5f6c443e4d24343e","subject":"Use rabbitmq parameters for tcp_listen_options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/96bd0f7ca64c95d96d64b55b5f6c443e4d24343e"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"subject":"Ensure the Aodh DB is synced","message":"Ensure the Aodh DB is synced\n\nWe were lacking the \"sync_db \u003d\u003e true\" parameter when calling aodh::api,\nso the database was not initialized, and API calls were failing.\n\nChange-Id: Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e\nCloses-Bug: #1855443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f64071f5a17dd3d66033edd5c3786d67459c293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f64071f5a17dd3d66033edd5c3786d67459c293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
