)]}'
{"id":"x%2Fpackstack~698541","triplet_id":"x%2Fpackstack~stable%2Ftrain~Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","project":"x/packstack","branch":"stable/train","hashtags":[],"change_id":"Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","subject":"Ensure the Aodh DB is synced","status":"MERGED","created":"2019-12-11 17:04:32.000000000","updated":"2019-12-12 11:49:08.000000000","submitted":"2019-12-12 11:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698541-1576151348357-fee75b61","meta_rev_id":"0dc389f452d1ba659b5e88643765ced9961a387a","_number":698541,"virtual_id_number":698541,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 11:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2019-12-12 10:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-12-12 10:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-12-12 10:53:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 10:50:04.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-12-12 10:53:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-12 11:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6edcdebda2c09dcc9574aa406add5839af30334","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 17:04:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"17756bb90f1b564d8dc9aafc06278c1e64723092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:56:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37f907add584fffa333ef083d738428 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a9fc69f24964fe89fef41516923e539 : SUCCESS in 4m 04s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/818c90fb32c244d98eb23883a095d484 : SUCCESS in 47m 06s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b9f2d43896f0460e8e8f4dcb06603247 : SUCCESS in 46m 47s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/87bd560f9bf24974983849693488765f : FAILURE in 12m 34s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb0c5dbf90894b38894e80adc0332f5f : SUCCESS in 46m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/914e6c9dc5a4460d95683524afa5ccfc : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"732a198d07f6037ae1b40b4ec59cf95b32e4c1ea","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 18:03:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5b47720eff075ea999bcded003bcf6f800316fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 18:57:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715f78ca3f8e451eb2a73a372f4a45e9 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aeefcfc80b7447f3a45eea9d8b5f3f85 : SUCCESS in 4m 46s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/1799373d05574e009c97222a033a3fc0 : SUCCESS in 49m 35s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/51a4265187174a91bde9f0b937da9a6e : SUCCESS in 50m 43s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/f8a026bd7f234686a7bafba418294b28 : SUCCESS in 41m 40s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/95f9230b4f974396a2d2179a332ca969 : SUCCESS in 44m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8137d54321c4fb28c34fc94de6c0d08 : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"878807dac008dbd63d4d9d384935a209530ac796","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-12 06:55:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7b877b6016228bc92846855b3619c528422e95c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 07:59:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec5dc360ac34f368ffa421fc7f342e3 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31f31bac6ff9424e9b22f29b74fc278f : SUCCESS in 5m 39s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/7ffab04be0744cbfab46a77d36c343fa : SUCCESS in 47m 27s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a2a2eb46e13d4280b79b486632406c50 : SUCCESS in 49m 00s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a268fb9ee3bf4259918214f824e408fb : SUCCESS in 51m 02s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1e66b7fa53614188b85bd23c206e4113 : FAILURE in 50m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8982ac4c55c4c3ca5a6b61e213a4800 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c81ef9a34743f02a322e7e8d5ee335b4ed8029d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-12 08:52:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"91dc9f44d330e2cdb675bee629e7b53fac533d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:38:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee7e3a1d39147a2bd423061178c242a : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a95694fbba784855a6cff4c8e4b4a3b3 : SUCCESS in 3m 01s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/83060febc9cb4f248022c1b4cda1b335 : SUCCESS in 42m 14s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b4084f8283f440fe954adf2b4f618dc3 : SUCCESS in 39m 35s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/0cb8b9f280684be1b1726ab4e1e0473b : SUCCESS in 42m 26s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f3b36f5f0a9e4c62908e631bf390d12f : SUCCESS in 45m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92bf0af715354822ba4e10344ba62c5e : SUCCESS in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"466a3da0cc9db96ab98d3c8469e16421ea4067ea","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-12 10:50:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"54233eed577caa6325e4e8b180605d80fded3c32","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-12 10:53:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d544984b356f11d1fdb5ad26236ed27860869391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 10:54:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"90a92dd8ca66a10e8208c0da645325fccfaa27b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:49:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6fe69bafd9478dab580e362f9fe330 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48fe65b11a984bb68d027b7d8b2cde97 : SUCCESS in 5m 03s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/b958446fea2e4875bb53f8a29c583ec2 : SUCCESS in 46m 59s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f4f488a133664e77aa483203e762552b : SUCCESS in 49m 28s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/dd865d29a66f4ee9ad8c60bc261de5f8 : SUCCESS in 46m 32s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/222cb45160d0437d985552397c37a129 : SUCCESS in 47m 57s","accounts_in_message":[],"_revision_number":3},{"id":"0dc389f452d1ba659b5e88643765ced9961a387a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c32f8cb9988b6509ad522e4f4aaff2a057c6fb88","revisions":{"4f7fe32b2fc44e956898426abe44b24da543abe1":{"kind":"REWORK","_number":1,"created":"2019-12-11 17:04:32.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/41/698541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/41/698541/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/41/698541/1"}}},"commit":{"parents":[{"commit":"f9431fe690cd61affaaf80de4a48803f85617115","subject":"Use https for cirros downloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f9431fe690cd61affaaf80de4a48803f85617115"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-12-11 17:04:32.000000000","tz":0},"subject":"Ensure the Aodh DB is synced","message":"Ensure the Aodh DB is synced\n\nWe were lacking the \"sync_db \u003d\u003e true\" parameter when calling aodh::api,\nso the database was not initialized, and API calls were failing.\n\nChange-Id: Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e\nCloses-Bug: #1855443\n(cherry picked from commit 1f64071f5a17dd3d66033edd5c3786d67459c293)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f7fe32b2fc44e956898426abe44b24da543abe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f7fe32b2fc44e956898426abe44b24da543abe1"}]},"branch":"refs/heads/stable/train"},"8d904b9109fed8a4f7e72c72333e53ee9b3e4f46":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-11 18:03:31.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/41/698541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/41/698541/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/41/698541/2"}}},"commit":{"parents":[{"commit":"07fb4b7fc449dd1eca718b89c8b46a06306a910f","subject":"Set unixsocket to undef in redis class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/07fb4b7fc449dd1eca718b89c8b46a06306a910f"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-12-11 18:03:31.000000000","tz":0},"subject":"Ensure the Aodh DB is synced","message":"Ensure the Aodh DB is synced\n\nWe were lacking the \"sync_db \u003d\u003e true\" parameter when calling aodh::api,\nso the database was not initialized, and API calls were failing.\n\nChange-Id: Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e\nCloses-Bug: #1855443\n(cherry picked from commit 1f64071f5a17dd3d66033edd5c3786d67459c293)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8d904b9109fed8a4f7e72c72333e53ee9b3e4f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8d904b9109fed8a4f7e72c72333e53ee9b3e4f46"}]},"branch":"refs/heads/stable/train"},"c32f8cb9988b6509ad522e4f4aaff2a057c6fb88":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-12 06:55:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/698541/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/41/698541/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/41/698541/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/41/698541/3"}}},"commit":{"parents":[{"commit":"e691b213f7fc30052952cf258d894e4e598827fe","subject":"Pin  non-OpenStack puppet modules in stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e691b213f7fc30052952cf258d894e4e598827fe"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-12-12 06:55:41.000000000","tz":0},"subject":"Ensure the Aodh DB is synced","message":"Ensure the Aodh DB is synced\n\nWe were lacking the \"sync_db \u003d\u003e true\" parameter when calling aodh::api,\nso the database was not initialized, and API calls were failing.\n\nChange-Id: Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e\nCloses-Bug: #1855443\n(cherry picked from commit 1f64071f5a17dd3d66033edd5c3786d67459c293)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c32f8cb9988b6509ad522e4f4aaff2a057c6fb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c32f8cb9988b6509ad522e4f4aaff2a057c6fb88"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
