)]}'
{"id":"x%2Fpackstack~698544","triplet_id":"x%2Fpackstack~stable%2Fqueens~Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","project":"x/packstack","branch":"stable/queens","hashtags":[],"change_id":"Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e","subject":"Ensure the Aodh DB is synced","status":"MERGED","created":"2019-12-11 17:04:57.000000000","updated":"2019-12-12 07:08:46.000000000","submitted":"2019-12-12 07:08:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698544-1576134526121-b3b91f8b","meta_rev_id":"c49907a50a4026e61466ff447357afa596550120","_number":698544,"virtual_id_number":698544,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 07:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2019-12-12 05:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-12-11 18:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2019-12-12 05:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 18:19:44.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-12-12 05:45:52.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-12 07:08:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ead00ae47eacc86c4c3e166512c80f4fb5eac77e","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 17:04:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"4288ff1cb98ef5cd9ef2a77259660ef7bc81594b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 18:07:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b6c875b6f7420b9d26f787c55ec3f5 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c03d0d5b429042e29fd52c3223331375 : SUCCESS in 3m 35s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/e617984cc2ba4af795a0690895b4c746 : SUCCESS in 51m 40s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4222ef4ec99249c694161f34394e8669 : SUCCESS in 44m 32s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ed9870703e1b42be8f462a8cb0083397 : SUCCESS in 56m 14s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/88a4dc07da2f41ee84babbf8adcf28a6 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":1},{"id":"91e36d81c9aca7131e4fa7f1845d01a3d5b90871","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-11 18:19:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84c7859ed77321d4326d6ee9076f580ed5a13c63","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-12 05:45:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"182adc8c4025bc583010066c1bbb8906d2602d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 05:46:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46152a4d89fe0ba85f34160735257fc23bf48aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 07:08:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977304e875ce48148da43bfd7fc3803a : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57d963a94d484638b0807a8674c1eb8e : SUCCESS in 3m 15s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/dd55c30ad4bd41b18f14d4bca6cfec6d : SUCCESS in 49m 54s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/91dcc9090a9b451686bd52db76489e44 : SUCCESS in 48m 59s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/af704b5721c444fa8de43e7c5df5c9ac : SUCCESS in 55m 13s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/25c1b48d2d314f84ae95dec40e133c29 : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c49907a50a4026e61466ff447357afa596550120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 07:08:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53617e23547c6a2987471500ae4666fb57b2642c","revisions":{"53617e23547c6a2987471500ae4666fb57b2642c":{"kind":"REWORK","_number":1,"created":"2019-12-11 17:04:57.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/44/698544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/44/698544/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/44/698544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/44/698544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/44/698544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/44/698544/1"}}},"commit":{"parents":[{"commit":"f21f84b066ab7af5b1df101159c0ba50d9316f8b","subject":"Reduce unnecessary log storage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f21f84b066ab7af5b1df101159c0ba50d9316f8b"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-12-11 12:15:58.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-12-11 17:04:57.000000000","tz":0},"subject":"Ensure the Aodh DB is synced","message":"Ensure the Aodh DB is synced\n\nWe were lacking the \"sync_db \u003d\u003e true\" parameter when calling aodh::api,\nso the database was not initialized, and API calls were failing.\n\nChange-Id: Idbc4f5a3310c0776ba4973b45dbaf554d8372b8e\nCloses-Bug: #1855443\n(cherry picked from commit 1f64071f5a17dd3d66033edd5c3786d67459c293)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/53617e23547c6a2987471500ae4666fb57b2642c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/53617e23547c6a2987471500ae4666fb57b2642c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
