)]}'
{"id":"x%2Fpackstack~698560","triplet_id":"x%2Fpackstack~stable%2Ftrain~Ic967d9446ef68a0257b9448bc8f3f3b664d776e0","project":"x/packstack","branch":"stable/train","hashtags":[],"change_id":"Ic967d9446ef68a0257b9448bc8f3f3b664d776e0","subject":"Set unixsocket to undef in redis class","status":"ABANDONED","created":"2019-12-11 18:03:18.000000000","updated":"2020-01-10 14:34:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44c6569eb5a29452d3e95699d86acf5805483649","_number":698560,"virtual_id_number":698560,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-11 19:05:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":-1,"date":"2019-12-11 18:23:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 18:23:24.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-12-11 19:05:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6f482e069e6084dd41e2674075a70dc58ec2123","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-12-11 18:03:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f592d31cf0aab6cd90a9cadc3b3bb611f727712","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-11 18:23:24.000000000","message":"Patch Set 1: Code-Review-1\n\nwe have puppet-redis pinned in RDO for train, i think we should just apply the same pin in Puppetfile","accounts_in_message":[],"_revision_number":1},{"id":"fc907ddf4ca468e2d974fc568b9e662249cc13af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:05:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb1603f6cc7b4bd59928e53e0257dc84 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/721a993eac1647a48f34023019fa4c15 : SUCCESS in 3m 06s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/4f26007a8d734ca8813bed3a0b45f676 : SUCCESS in 42m 51s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/44fbed38fadd4cb483406fd9afcfc6c1 : SUCCESS in 37m 56s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/c7e530a91fc24cc389ac12ea31a31932 : SUCCESS in 50m 33s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6592f73f8803413eb14587bde8bb71a0 : SUCCESS in 59m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6dd609dce924dadb77b1cd8cb2eaf77 : SUCCESS in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44c6569eb5a29452d3e95699d86acf5805483649","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-01-10 14:34:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07fb4b7fc449dd1eca718b89c8b46a06306a910f","revisions":{"07fb4b7fc449dd1eca718b89c8b46a06306a910f":{"kind":"REWORK","_number":1,"created":"2019-12-11 18:03:18.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/60/698560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/60/698560/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/60/698560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/60/698560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/60/698560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/60/698560/1"}}},"commit":{"parents":[{"commit":"f9431fe690cd61affaaf80de4a48803f85617115","subject":"Use https for cirros downloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f9431fe690cd61affaaf80de4a48803f85617115"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-11-27 09:25:43.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-12-11 18:03:18.000000000","tz":0},"subject":"Set unixsocket to undef in redis class","message":"Set unixsocket to undef in redis class\n\nAfter [1], redis class does not longer accepts boolean as value for\nunixsocket, so let\u0027s undefine it. Note that this has the effect of\nenabling unix socket in redis but I don\u0027t see a way to get it disabled\nafter the patch.\n\nTo minimize risk, i\u0027m setting permissions of socket to 0700.\n\n[1] https://github.com/voxpupuli/puppet-redis/commit/502ff38c8c035fa143e32dd4ee61b3ec68db461c\n\nChange-Id: Ic967d9446ef68a0257b9448bc8f3f3b664d776e0\n(cherry picked from commit 4a1fd433e8482f58edd741ccfc70ec9fa473806b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/07fb4b7fc449dd1eca718b89c8b46a06306a910f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/07fb4b7fc449dd1eca718b89c8b46a06306a910f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
