)]}'
{"id":"x%2Fpackstack~699334","triplet_id":"x%2Fpackstack~stable%2Ftrain~Ie53a8cb5517651487436e24bf60f51bc531f16d6","project":"x/packstack","branch":"stable/train","topic":"rdo-centos8","hashtags":[],"change_id":"Ie53a8cb5517651487436e24bf60f51bc531f16d6","subject":"Add jobs for CentOS 8","status":"MERGED","created":"2019-12-17 05:01:32.000000000","updated":"2020-06-30 13:52:54.000000000","submitted":"2020-06-30 13:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699334-1593525174635-b1cf2b13","meta_rev_id":"1cd3ba3b96877c79bcdc6c2f3f38a799cc240973","_number":699334,"virtual_id_number":699334,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 13:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-06-30 12:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 12:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 12:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 12:13:17.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-06-30 12:42:20.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-30 13:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44b22cc7d9eae187a336e902beea62d4b8530323","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-17 05:01:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07e907a24bd4b16e066d169aa0d1f81a1cfed796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 07:10:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421a4b5cc64449eea73233b44008e3fa : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/654368b6bf224d03aa5d3204e65c63d4 : SUCCESS in 3m 17s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c6ef664d038b4bfa888550c32a695e75 : TIMED_OUT in 2h 02m 01s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/acbef7005cd745ad971e48dff01e68ab : TIMED_OUT in 2h 01m 29s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1123487ee1594d8ba58a7c6646c94548 : TIMED_OUT in 2h 01m 36s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1b5e2bf3604e88984c891dfed67bff : SUCCESS in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7836e732e946327c4899255a72b5e11ba8b712aa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-17 08:35:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b2c3a9c3dde47182725eed56b5e47c15ae1ad26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 09:25:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2a70094ea94ae9aaa0196409c2c1ae : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7db4ae38cb9844b39b8f37ae5a0f3c2b : SUCCESS in 3m 41s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ad325d7c15724f6f9afd183b244cab26 : SUCCESS in 46m 22s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b30b8c6a39df4dc18b6e4a0230cb00b2 : SUCCESS in 44m 38s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8129daa55f1144b4bd8994ece77e8d04 : SUCCESS in 44m 42s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49b97da51b047a287e4e34a2ab11808 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05e6786c8341bcc2050a3cf41516458cd09cfc75","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-17 19:07:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab4f8d3641558dbd99a380bcd5232e1eb606ba00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:09:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9557e3ee297c44cf9bf85da23e107a12 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/131e0a86dd234eb4956f5cc1aaf50042 : SUCCESS in 3m 53s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/c14b0f489f014bc48065f6d76c6ac5fb : SUCCESS in 41m 00s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2d54cec279b848a48d7159b34806fa4f : SUCCESS in 42m 05s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/683e09dc94ed40708dbe42049362c069 : SUCCESS in 46m 50s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1229acee3a5d4e83a0f86b5a23499b20 : SUCCESS in 51m 14s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9409ce8dc2a1446190cc9020653bdf92 : SUCCESS in 42m 14s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1b1cf0dc5a424babb9f3338021f52556 : SUCCESS in 48m 40s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4b4b830f97844db0afd5d9a8212ec219 : FAILURE in 19m 19s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3d97d46066684954bc52119085650df6 : POST_FAILURE in 26m 26s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c87509d987a547d0866a3064aaf988ff : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e8203e3d0d9404a10e267b6c93f40fe49fbf406","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-17 20:30:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e56490a75f624c4ae9a4353031789b9cec00c49f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-17 20:32:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d69fe4a3682c349bae794aba7bfc27571a53bb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:31:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b714be03636041dfb5c7a67aa462d909 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35efcb5cd0c846b6a97391ccac8dbbfd : SUCCESS in 6m 03s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/65463828ae3d410f90729fd10d5c35b2 : SUCCESS in 42m 16s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5732ef48cf09414da7c6654f88f8014f : SUCCESS in 45m 07s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/9e09a1de831647b1be5820a4d6fc6d06 : SUCCESS in 53m 35s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7af1c01fd90d4ad690c93b7db42c8c92 : SUCCESS in 49m 42s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8dba9be9dc8c465a81adf38de60498cf : SUCCESS in 46m 46s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0353ac8e12cf4260818967597f926c4f : SUCCESS in 52m 46s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1c3a35749ef04f8988709a295e73677e : SUCCESS in 56m 57s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3b02eac050344bcf835463b169868e62 : FAILURE in 35m 22s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a83f469416c14886886b811f48c959a4 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"26b079b8227bb018425ebe0aaa51c317d55a89d3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-17 22:22:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"053dbea58134dada53d0b5a927b4c563c9ae6cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:38:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3ab6e7479a4404809dba5bce84ce00 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49faa14b7b984a9587f043f5e63dd1d6 : SUCCESS in 5m 53s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/1e11f09bcd42439299684bfe180fcfe3 : SUCCESS in 1h 12m 13s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dc7355efb06846c09d4b0435d37609b6 : SUCCESS in 45m 22s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/6f7247ea71cd439c985d51156a7b060c : SUCCESS in 47m 25s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/618c38be7791409bb993f3021b67879a : SUCCESS in 45m 15s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2fa28883450145b187b8711cba72d575 : SUCCESS in 44m 21s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d4ee4decb5684bedab2a7f1f3a707f38 : SUCCESS in 45m 36s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b9d8cf08580a4ea68cb2972e23080bc6 : SUCCESS in 47m 04s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ac9bbbee5a124555997634cced49a5f1 : POST_FAILURE in 36m 15s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d890b5ff3d944faa92a2e8544f80da08 : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52e7cc0e2470bbdb4b56e9903b22c8bdef3be4c3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 08:36:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1f4670433fda571aec58831727a69388e46a5058","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 08:52:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"90ecc7b15fd4898af4a9405f5e607ab89f1a99d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 10:07:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b6d1b132f24020b735291cc06ef327 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcd0cd63ea044e51b444bfced22e507f : SUCCESS in 5m 01s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/fa56395c33934f179302ed3bc614c672 : SUCCESS in 1h 00m 34s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/93d12be20c2d4303963080fdb6b07b9a : SUCCESS in 45m 31s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/f1e6469843a5478ab52f70987294a165 : SUCCESS in 46m 39s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e90131ebcea84783aa0d17be1fb9a535 : SUCCESS in 1h 01m 54s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e8c028f0a2d94fb9839194a315f28dda : SUCCESS in 56m 22s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/49916751195f496b802280c2691858c2 : SUCCESS in 49m 28s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/93371ec8be28428c86d8fb84d3cf6c60 : SUCCESS in 50m 26s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/22d5a60b83c14a4587953319b8e20439 : FAILURE in 42m 19s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b01a840d1bb4ee49a12444000f83d2e : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1f9def6f0ebe6da2f2bf1fb7bc238cea8e7dacba","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 10:57:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a85261ae30f3474d92d286167505414395bafc1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 11:50:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05033d368a714611a7d5da23cb287ad8 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66e9cb98b0254861b7f6b711f565166c : SUCCESS in 3m 41s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/22f932eb97fa4a479f8a262ce870745c : SUCCESS in 42m 09s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3a9ce8a2632b467f878d0807dd30d2ab : SUCCESS in 39m 44s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/55008928879d4975b9991521392c6494 : SUCCESS in 46m 47s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c62e8265554848aebad683e230f48e4f : SUCCESS in 44m 39s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9e6546e60d424006a9a8496f46e7003b : SUCCESS in 42m 54s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7358fd8d9e034e26ab9e2321137202a9 : SUCCESS in 46m 17s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a58080ffa4154a7bb3b547ef3a17c75a : SUCCESS in 47m 11s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb99015f884f447a8228bf01f06768dd : POST_FAILURE in 38m 30s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d485a4efaef41eda9454954545fa091 : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2289d27eade38d3a9f7fb298879b45e51e3e6ae4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 14:05:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"618a1b956153e8d9c68dabffd95c1ffb8423f053","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 14:32:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ba301a0ee7160bd20415605ca54c17a09e3df9a0","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 14:50:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8753d9b71eeac9bdb91dc2ed4125d37a7e4150d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 15:57:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bef204893646a6b0d37bf8296d81a8 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/777ef87dee424a4682feb2b74807696e : SUCCESS in 5m 34s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/94abcfc8dc554651848396add64ae998 : SUCCESS in 37m 37s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d7a16ac4ea1a48a9bbd11556b881b138 : SUCCESS in 49m 15s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/71f27c3770164019a964fe575bb24bac : SUCCESS in 56m 29s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a1aa4751def04eefa42578d6996a0018 : SUCCESS in 59m 09s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9dd8b8ac4b2e405cb6646a0306370d7c : FAILURE in 34m 16s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/462be857b02e41a9ab32cce8631dc701 : SUCCESS in 48m 42s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9e09a4940767453fb4b55196becea138 : SUCCESS in 51m 01s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2e4f614528714918baa632870075fe7c : POST_FAILURE in 47m 54s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83e4ff9ce71a42a88a1accf005251887 : SUCCESS in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c93971f35257ed4bf7948e54f5df9c7196bb2eb6","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 16:09:10.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d721e4472f5bd067be241b061168f217ca321062","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 16:43:08.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f63fb6e8b601ba017e35995e3c2601f39ff3fd49","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-18 16:59:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3252428ac1e8074b9061dc50e369a10d476808fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 18:02:31.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d36c18dfe4d49c8be9bbd2c96aa99e3 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78ab977bedc0404b9f43c5143f84a42a : SUCCESS in 3m 37s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/87f300dfc35b4a49a70f7d97d34d9037 : SUCCESS in 48m 09s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/80b2d9c0b7cd4977843910d10c6451a5 : SUCCESS in 45m 55s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/af3b1cc7246e439a884fd21984e9c535 : SUCCESS in 52m 52s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/90127c9da9ee4198bbc422d83be45777 : SUCCESS in 56m 21s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/00616ab70bbd4c4994d25d180d55c1ae : SUCCESS in 46m 11s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bd8826b3479143b3951470a4e3cdfa22 : SUCCESS in 51m 16s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8608d1c7e3fe4f2ebd01c4633b880f42 : FAILURE in 50m 05s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/668d8069b03447108b92ced16c8bbfb5 : POST_FAILURE in 50m 56s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a23dbe0574c2441b9d9d46dc66e34d29 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5cd45e6b2567bd8910902b71f27f217ec0d2312b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-19 09:46:10.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"21aa71f035838d36bcaec25fb1b3ff59920cfcc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:45:22.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f152b22f2da64571a8140a391f902fed : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa5978230fa24d5fa73fbfc87553e622 : SUCCESS in 3m 21s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/51cc16f27f574bb19f727eafcdabc1c7 : SUCCESS in 39m 28s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f91bd837af2248e7ac6dc8155363ebdb : SUCCESS in 40m 31s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/9a72e344d10c4a139625509ae2a2b485 : SUCCESS in 50m 36s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/91654253716a455fb021dfec44ce11b5 : SUCCESS in 41m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ee0fcdc762de41e3a283841213b3881a : SUCCESS in 42m 22s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8b1e228582db400b80ee76fddd1909ac : SUCCESS in 44m 44s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3d41d14710b64166a6dc0c0588238c0f : SUCCESS in 47m 17s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a05a27dab0564301a0dd14f862f368cd : SUCCESS in 52m 51s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef2185c1d8e440e90c9d47d7deca721 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"34fdd7e9c83e2203cc631476f5d1e128691eebef","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-29 13:35:36.000000000","message":"Patch Set 16:\n\nAre we still interested in this?","accounts_in_message":[],"_revision_number":16},{"id":"ddc2d45e73398eebe8611e355596b2e3ef92a9ec","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 13:38:01.000000000","message":"Patch Set 16:\n\n\u003e Are we still interested in this?\n\nI\u0027d say so","accounts_in_message":[],"_revision_number":16},{"id":"796f87660292c19f8aa698d444a9d36c6721d7f7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 13:38:12.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"0d06e00c37c61c13791fd8e5ff03647e9922c0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 14:43:03.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60abd3a143804fe79c0b554ecf0caf80 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e3768cf6fa847a1b8e15f087b6b2594 : SUCCESS in 9m 43s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/6c19ecad9e27467c9c4975f55d38ac10 : SUCCESS in 39m 33s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/afe778338bce48e6b372c2d527c3244a : SUCCESS in 52m 22s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ce4ed5aa056e450291171b917ba4e0b6 : SUCCESS in 56m 14s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ce39b56c88664cd9b5b73de4b04f8521 : SUCCESS in 58m 32s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/48efe77dc3d94a5db6a43f3ce6e61f08 : POST_FAILURE in 11m 12s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aec8744d993a4f40b00abb0ca06d690b : POST_FAILURE in 11m 24s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e6ac6fdae9b44bab0a116065b8620a0 : POST_FAILURE in 6m 45s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c8135acbbb5446a1a19df97962c0fda0 : POST_FAILURE in 7m 39s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf2dbbdfa13412fb0f5bba531b07438 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"764f7a684d9454fc48f1676a77103d0480215753","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 09:54:04.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"7b3dbb8590b9c8377d813d78c771853087f68e5a","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 09:59:59.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"5ee85ec550b87dcee246c92bedb2e4f71a03e916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:55:20.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94644fa68f32402eb24e72a987dc7b6f : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a597e9d23724b859079ac3f76e78bb8 : SUCCESS in 4m 14s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/82a8d3679f9742869213cd7508ac1d5c : POST_FAILURE in 14m 08s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4419f7a4ed284c94aa99ad84f28b4b50 : SUCCESS in 43m 54s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/d2bf378e209f44c0a25b7f823bbbb3d6 : SUCCESS in 44m 17s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/136caefdddbf4a489210288f77a4c0ad : SUCCESS in 48m 56s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5b7c3a54f9ee4edbbb644dced6afc624 : SUCCESS in 40m 07s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ddc59e4045484056b15481e63fc46bb1 : SUCCESS in 41m 12s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/911b25237e014445b4b0d81005bb35c4 : SUCCESS in 53m 51s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f7dd9cdea9474768afff2fb6b0c33b3b : SUCCESS in 36m 51s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04b2cf62a00f41499c7319f16880fd50 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6d38ef457ab3d086c132816b1ebd4ce05f563a5b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 11:01:53.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"61b5c54df8aaedb6c71ea42f0182d4534b8e56d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 11:59:55.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e8389f1a46492985670aa5f6695f61 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/134d92e5149f4bcf86528661e096c406 : SUCCESS in 4m 37s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/5b0a4409930f40788fe9dd300caac5aa : SUCCESS in 41m 01s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6f844568d11f487597f0238a0c719ba2 : SUCCESS in 27m 12s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/f81b5a13513b4d128051376a9699f3c8 : SUCCESS in 46m 46s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8279fc043fb7459a8296f1d49b723506 : SUCCESS in 53m 51s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b37b7a0deea14a3ea5d4bb453fde7561 : SUCCESS in 47m 35s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3dbb8f3676034086a58ff4b32fb1f542 : SUCCESS in 45m 44s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a5e90463554e413bb4e7de65798b5e8a : SUCCESS in 32m 01s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9b9809c172434dfe9241d39f5d299d81 : SUCCESS in 48m 19s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a73b076ef6f4ec782d2e833b1280ff7 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7f866dfbc7d87840f8a7445615373cbc67a86e21","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-30 12:13:17.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"98074a9325695261a67a7d43e634881dba369110","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 12:42:20.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"c10a3f0421a7750ead4d208b25fabb142311b88d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:42:35.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"d7de5915cc9d4de6413adb769d3b3eb8b8ffac37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:52:53.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a6d37b912e46ba94b6f2937f7cffdf : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/362a26660ddd4c7e89fdecbb3df808f2 : SUCCESS in 5m 30s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/a703d57a24ec47a3beab447fac9cf7a5 : SUCCESS in 50m 33s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3362b2120b464f3d9f5ec5c972221619 : SUCCESS in 1h 04m 32s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/09486cbbf04d431d8d2b2109b2e67a6a : SUCCESS in 1h 08m 03s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1d3e1130003d427fbeddbcd85d49120a : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":17},{"id":"8ddd459010a44a22c1339eaff394f3ce1c190948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:52:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"956eda4a5590cdca82a21a46d9be0c5899581627","revisions":{"95dfd70f75e64af506e9193dbe9f4653e9ec7ee0":{"kind":"REWORK","_number":1,"created":"2019-12-17 05:01:32.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/699334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/1"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-17 05:01:11.000000000","tz":330},"subject":"[DNM] Test CentOS 8","message":"[DNM] Test CentOS 8\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 0e6b8c89d1b4f8bb5b8eeb48fdf44070223b7007)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95dfd70f75e64af506e9193dbe9f4653e9ec7ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95dfd70f75e64af506e9193dbe9f4653e9ec7ee0"}]},"branch":"refs/heads/stable/train"},"eb84ba54d7a25f50bc6ec863fcf68030d07dbc26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-17 08:35:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/699334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/2"}}},"commit":{"parents":[{"commit":"bfd32144835e1b43a760d9a4a498c4f21eae49db","subject":"Use rabbitmq parameters for tcp_listen_options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bfd32144835e1b43a760d9a4a498c4f21eae49db"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-12-17 08:35:25.000000000","tz":0},"subject":"[DNM] Test CentOS 8","message":"[DNM] Test CentOS 8\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 0e6b8c89d1b4f8bb5b8eeb48fdf44070223b7007)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb84ba54d7a25f50bc6ec863fcf68030d07dbc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb84ba54d7a25f50bc6ec863fcf68030d07dbc26"}]},"branch":"refs/heads/stable/train"},"76023420dfc0070c8062e5a83173e1e9c1519bf1":{"kind":"REWORK","_number":3,"created":"2019-12-17 19:07:31.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/3"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-17 19:07:08.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/76023420dfc0070c8062e5a83173e1e9c1519bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/76023420dfc0070c8062e5a83173e1e9c1519bf1"}]},"branch":"refs/heads/stable/train"},"c1e7a864ad7e7af874fb0c4718702686c20182f8":{"kind":"REWORK","_number":4,"created":"2019-12-17 20:30:39.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/4"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-17 20:30:19.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c1e7a864ad7e7af874fb0c4718702686c20182f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c1e7a864ad7e7af874fb0c4718702686c20182f8"}]},"branch":"refs/heads/stable/train"},"cb808993f0874481d6500cfe716b56a9a13b75cb":{"kind":"REWORK","_number":5,"created":"2019-12-17 20:32:24.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/5"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-17 20:32:13.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cb808993f0874481d6500cfe716b56a9a13b75cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cb808993f0874481d6500cfe716b56a9a13b75cb"}]},"branch":"refs/heads/stable/train"},"cdcbe52cb604852420651550b08acb8e0ce16f96":{"kind":"REWORK","_number":6,"created":"2019-12-17 22:22:22.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/6"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-17 22:22:10.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cdcbe52cb604852420651550b08acb8e0ce16f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cdcbe52cb604852420651550b08acb8e0ce16f96"}]},"branch":"refs/heads/stable/train"},"a2bf319fc415edfd7db9fc092210b12111de4518":{"kind":"REWORK","_number":7,"created":"2019-12-18 08:36:03.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/7","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/7"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 08:35:47.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a2bf319fc415edfd7db9fc092210b12111de4518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a2bf319fc415edfd7db9fc092210b12111de4518"}]},"branch":"refs/heads/stable/train"},"cf4bff388f0ec39117f5da4195a521b2f14c8a55":{"kind":"REWORK","_number":8,"created":"2019-12-18 08:52:52.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/8","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/8"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 08:52:38.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cf4bff388f0ec39117f5da4195a521b2f14c8a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cf4bff388f0ec39117f5da4195a521b2f14c8a55"}]},"branch":"refs/heads/stable/train"},"ba71bc81f9d6b36e001b1abe88ab562b08b1ff5a":{"kind":"REWORK","_number":9,"created":"2019-12-18 10:57:51.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/9","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/9"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 10:57:40.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ba71bc81f9d6b36e001b1abe88ab562b08b1ff5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ba71bc81f9d6b36e001b1abe88ab562b08b1ff5a"}]},"branch":"refs/heads/stable/train"},"44103290ca8f23ccf889abe23a864d31d278d46b":{"kind":"REWORK","_number":10,"created":"2019-12-18 14:05:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/10","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/10"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 14:04:55.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/44103290ca8f23ccf889abe23a864d31d278d46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/44103290ca8f23ccf889abe23a864d31d278d46b"}]},"branch":"refs/heads/stable/train"},"484856f3c384819ed640fc3484e3349c47b216b0":{"kind":"REWORK","_number":11,"created":"2019-12-18 14:32:53.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/11","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/11"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 14:32:38.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/484856f3c384819ed640fc3484e3349c47b216b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/484856f3c384819ed640fc3484e3349c47b216b0"}]},"branch":"refs/heads/stable/train"},"0bbd349fadd024092f7a93f74c065f68696363e9":{"kind":"REWORK","_number":12,"created":"2019-12-18 14:50:52.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/12","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/12"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 14:50:34.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0bbd349fadd024092f7a93f74c065f68696363e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0bbd349fadd024092f7a93f74c065f68696363e9"}]},"branch":"refs/heads/stable/train"},"eae6965d764f39129de2b5a9b7f6c396f13f5952":{"kind":"REWORK","_number":13,"created":"2019-12-18 16:09:10.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/13","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/13"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 16:08:59.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eae6965d764f39129de2b5a9b7f6c396f13f5952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eae6965d764f39129de2b5a9b7f6c396f13f5952"}]},"branch":"refs/heads/stable/train"},"ece570dcedb7c0cc481ee796b6975d4561ca80cb":{"kind":"REWORK","_number":14,"created":"2019-12-18 16:43:08.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/14","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/14"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 16:42:57.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ece570dcedb7c0cc481ee796b6975d4561ca80cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ece570dcedb7c0cc481ee796b6975d4561ca80cb"}]},"branch":"refs/heads/stable/train"},"c0556a208f2b9347019918005cfb2a2456e704a6":{"kind":"REWORK","_number":15,"created":"2019-12-18 16:59:34.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/15","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/15"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-18 16:59:22.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c0556a208f2b9347019918005cfb2a2456e704a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c0556a208f2b9347019918005cfb2a2456e704a6"}]},"branch":"refs/heads/stable/train"},"6981676fbf63891f4dd2e2285842ef0e5dc8df6a":{"kind":"REWORK","_number":16,"created":"2019-12-19 09:46:10.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/34/699334/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/16","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/16"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-19 09:45:54.000000000","tz":60},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6981676fbf63891f4dd2e2285842ef0e5dc8df6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6981676fbf63891f4dd2e2285842ef0e5dc8df6a"}]},"branch":"refs/heads/stable/train"},"956eda4a5590cdca82a21a46d9be0c5899581627":{"kind":"REWORK","_number":17,"created":"2020-06-30 09:59:59.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/34/699334/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/34/699334/17","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/34/699334/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/34/699334/17"}}},"commit":{"parents":[{"commit":"372d12079b34eef7df0aba3f078d9da240befbe9","subject":"Make sure rabbitmq can read CA certificate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/372d12079b34eef7df0aba3f078d9da240befbe9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-12-10 03:40:40.000000000","tz":330},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-06-30 09:59:56.000000000","tz":0},"subject":"Add jobs for CentOS 8","message":"Add jobs for CentOS 8\n\nAs part of adoption of CentOS 8 in RDO, this patch is adding support to\ntest with it:\n\n- Adapt run_tests.sh for CentOS8.\n- Add the option to have different versions of puppet modules (needed\nfor puppet-mysql until we can update mariadb in CentOS7.\n- Add centos8 playbooks and jobs definition.\n\nChange-Id: Ie53a8cb5517651487436e24bf60f51bc531f16d6\n(cherry picked from commit 9ddd5469d64a0459b98927efcf370547c69a5c0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/956eda4a5590cdca82a21a46d9be0c5899581627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/956eda4a5590cdca82a21a46d9be0c5899581627"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
