)]}'
{"id":"x%2Fpackstack~700173","triplet_id":"x%2Fpackstack~master~I9108299b2fffc9b4b84f81065faf18d5521004b8","project":"x/packstack","branch":"master","topic":"DNM","hashtags":[],"change_id":"I9108299b2fffc9b4b84f81065faf18d5521004b8","subject":"DNM Test new centos8 deps repo from trunk.r.o","status":"ABANDONED","created":"2019-12-20 13:34:21.000000000","updated":"2020-06-02 09:25:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d81198fc0c654cc50609ce1d7db57e4035093c8a","_number":700173,"virtual_id_number":700173,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-20 14:09:01.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-20 14:02:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 14:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"902ccfac4ba13b1e34b34a1d5bb9757a2f687b7c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-20 13:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"242bcf658d882cf691e44ebecf014ddb5a0314b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 14:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fd72530d01499e83f6a9269ca43180 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e8625e6631a46328fe40bce4dc196dd : SUCCESS in 4m 01s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8ad94eb90ce7487e9d072dd0eb8a6c9c : FAILURE in 20m 06s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1e6d1f6ec04645cd9b243f0f6b6e98d5 : FAILURE in 21m 24s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1e146e5c5b9740d39ecc5f1bca056985 : FAILURE in 24m 23s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8792418ee7df4be48faf49417afbd7cf : FAILURE in 22m 13s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0a1d9ce243045a280fd5f77430a5152 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98cd723a18515295118de78596106bdbfd3f9b43","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-12-20 14:09:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35553ba4b22c7ce32e85f3d1f7b3d4d51cc18325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 15:17:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba462056e824257bbfa933d4f67a470 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5779d72383a242549bf06beccb495508 : SUCCESS in 3m 35s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d2ce5c24a6704b11a213feca22ba6e9e : SUCCESS in 43m 20s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5ac0cc835d7d46a0aaa0ed632983999e : SUCCESS in 41m 13s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/245eee4cbcf845d499e801e684222a01 : SUCCESS in 1h 04m 20s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d61a49e8c7d84cf2a1f57abb6a610d76 : SUCCESS in 50m 02s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9abaf8c374ea4bb5baec9ff2f3ae177d : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efc89ef7b8a74ae503889d81c18badd8f33afa89","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-02 09:25:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20a24fc76a6eb3eb50a1464441b426a0b4ef5c17","revisions":{"20a24fc76a6eb3eb50a1464441b426a0b4ef5c17":{"kind":"REWORK","_number":1,"created":"2019-12-20 13:34:21.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/73/700173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/700173/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/700173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/700173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/700173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/700173/1"}}},"commit":{"parents":[{"commit":"a344b9cd369a65a6ad22d50acfcb010ceadd65f2","subject":"Add jobs for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a344b9cd369a65a6ad22d50acfcb010ceadd65f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-20 13:32:45.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-12-20 13:33:28.000000000","tz":60},"subject":"DNM Test new centos8 deps repo from trunk.r.o","message":"DNM Test new centos8 deps repo from trunk.r.o\n\nChange-Id: I9108299b2fffc9b4b84f81065faf18d5521004b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20a24fc76a6eb3eb50a1464441b426a0b4ef5c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20a24fc76a6eb3eb50a1464441b426a0b4ef5c17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
