)]}'
{"id":"x%2Fpackstack~701532","triplet_id":"x%2Fpackstack~master~I8d226aa2218eafc07119f88eca5be24c3514b717","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I8d226aa2218eafc07119f88eca5be24c3514b717","subject":"Do not create heat_stack_owner role in keystone.pp","status":"MERGED","created":"2020-01-08 11:15:13.000000000","updated":"2020-01-09 09:40:02.000000000","submitted":"2020-01-09 09:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701532-1578562802637-1b68ee4e","meta_rev_id":"0513a433f92088812114b8ae6ed8ffc961f9e12b","_number":701532,"virtual_id_number":701532,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 09:40:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-01-09 07:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-09 08:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-09 08:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 07:11:31.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-01-09 08:35:20.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-01-09 09:40:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9db442036057e664b3913f45b18bd5ebae574a01","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-01-08 11:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822877bff4e673a13f58fbcaa653872db58f52f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 12:09:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787c3e262d1e4b85a3f04dac88b734e7 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eefe1b960a9443b3b797d8f4851df68c : SUCCESS in 3m 41s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/b4807367866541a7a21c96f658c73ee0 : POST_FAILURE in 11m 23s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f4f58f7b108f4b42a9e9651b34b9c03d : POST_FAILURE in 7m 19s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/0c39efce74a1453188b2e88e7002e455 : POST_FAILURE in 7m 33s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a8cfb502513544bab81be7ab339f0cdf : POST_FAILURE in 8m 38s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/05a1b7b619374ccc89f75de6c14b54e5 : SUCCESS in 42m 30s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/539f41e9deb24c14940e6ada801dda79 : FAILURE in 22m 37s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6daf012eaea546d7846c8a786325589e : FAILURE in 11m 41s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a44fc52735984eb299d490d6423f14e7 : FAILURE in 17m 13s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23e56219e042401d982a5dd191197a0f : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cea08076f8652987bf45d9321924eb2ac72a6e19","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-01-08 17:19:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a71f603a057a183932203f2fd84ec285b4b116cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 18:52:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7812b85d5164423893e8c7d60736e3b3 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d55a612024ae4d6b8778b7d95e3ded10 : SUCCESS in 2m 47s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/e9e1dc7c8e644cf58e9349b9c9a9b41d : SUCCESS in 47m 01s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a5826d06d641487b843581100d2085ab : SUCCESS in 50m 22s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a966c6d973434da8a89016d33f432560 : SUCCESS in 1h 03m 18s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ea3d70b1d5fd492ba6c722205c798dd2 : SUCCESS in 1h 02m 17s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6bb83e5c5b034818bd7278fc02ef13ff : SUCCESS in 1h 06m 13s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fbced36047fc426abc82175b889b9e06 : SUCCESS in 48m 29s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ff8387c12db44672911d267be0792478 : SUCCESS in 1h 07m 58s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b951a3b6fcee42eb9ae685d46e2728dc : SUCCESS in 1h 04m 40s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e5b34eb29046a7aca307fca0ef6dc7 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0eb6cf372c38963d7ffa2d2be00874c891d8e4f2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-09 07:11:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe6a31beadb7edb1053b26580fd80175f6917e55","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-01-09 08:35:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75d78d9c291fa85e74a4aae9c1470834dcccdff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 08:35:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d07847c035c8103d243748226dc5b47a6f6a7cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:40:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4108084b614049c489750dc486b6628e : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbee5e10e1a442ca94f2fa350b69b67f : SUCCESS in 3m 29s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/24bbe52776854d7abb99b08ebcd38fea : SUCCESS in 46m 56s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/49bb16eb978e40e6b62e519c5335188c : SUCCESS in 52m 59s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a39933983f01443c9b2100021c6c2d2a : SUCCESS in 58m 08s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d6dedd90355f4db68f38a8fa954e9107 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6668f9878792bb0efe72dcc92aac4d60c1467796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:40:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c86f30c1dd7017598ccb8903577e8a1101df0f72","revisions":{"dc08b795ec26d73ea0167ebc89e4eca828a571a3":{"kind":"REWORK","_number":1,"created":"2020-01-08 11:15:13.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/32/701532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/701532/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/701532/1"}}},"commit":{"parents":[{"commit":"a344b9cd369a65a6ad22d50acfcb010ceadd65f2","subject":"Add jobs for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a344b9cd369a65a6ad22d50acfcb010ceadd65f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-08 11:11:06.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-08 11:15:10.000000000","tz":60},"subject":"Do not create heat_stack_owner role in provision.pp","message":"Do not create heat_stack_owner role in provision.pp\n\nAfter [1] it\u0027s created in keysone::bootstrap when\nkeystone::roles::admin::admin_roles includes it as done in packstack\n[2].\n\n[1] https://review.opendev.org/#/c/692664\n[2] https://github.com/redhat-openstack/packstack/blob/a344b9cd369a65a6ad22d50acfcb010ceadd65f2/packstack/puppet/modules/packstack/manifests/keystone.pp#L53\n\nChange-Id: I8d226aa2218eafc07119f88eca5be24c3514b717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dc08b795ec26d73ea0167ebc89e4eca828a571a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dc08b795ec26d73ea0167ebc89e4eca828a571a3"}]},"branch":"refs/heads/master"},"c86f30c1dd7017598ccb8903577e8a1101df0f72":{"kind":"REWORK","_number":2,"created":"2020-01-08 17:19:17.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/32/701532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/701532/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/701532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/701532/2"}}},"commit":{"parents":[{"commit":"a344b9cd369a65a6ad22d50acfcb010ceadd65f2","subject":"Add jobs for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a344b9cd369a65a6ad22d50acfcb010ceadd65f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-08 11:11:06.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-08 17:16:59.000000000","tz":60},"subject":"Do not create heat_stack_owner role in keystone.pp","message":"Do not create heat_stack_owner role in keystone.pp\n\nIt\u0027s created in heat::keystone::auth and after [1] it\u0027s complaining of\nduplicate declaration.\n\nAlso pinning gettext as version 3.3.0 is not compatible with the ruby\nversion in CentOS7.\n\n[1] https://review.opendev.org/#/c/692664\n\nChange-Id: I8d226aa2218eafc07119f88eca5be24c3514b717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c86f30c1dd7017598ccb8903577e8a1101df0f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c86f30c1dd7017598ccb8903577e8a1101df0f72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
