)]}'
{"id":"x%2Fpackstack~701711","triplet_id":"x%2Fpackstack~stable%2Fqueens~I8d226aa2218eafc07119f88eca5be24c3514b717","project":"x/packstack","branch":"stable/queens","hashtags":[],"change_id":"I8d226aa2218eafc07119f88eca5be24c3514b717","subject":"Pin gettext to \u003c 3.3.0","status":"MERGED","created":"2020-01-09 11:06:32.000000000","updated":"2020-01-09 13:26:08.000000000","submitted":"2020-01-09 13:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701711-1578576368387-af565c7f","meta_rev_id":"b4cce9e49e3a57bc7d15c0fd211ebd567e55e0b0","_number":701711,"virtual_id_number":701711,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 13:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-01-09 11:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-09 11:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-01-09 12:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 11:10:15.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-01-09 12:18:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-01-09 13:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce4af43150d19c5893308666da0b8fd876535f72","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-01-09 11:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1406962d5e11e5e406b3321fd63be20876d849c8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-01-09 11:10:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b01c1276b8a84fee2e881444bf02842b157c39f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-09 11:39:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc3bb9d8f6002fc4748b920dabf10eb263aa85bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:16:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979e050e32394109b0acc4c9f71cdd24 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ddefc8f279c4a039df51442868a180c : SUCCESS in 4m 19s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/fcf5758f2f0e4e00a5e7de5a864db584 : SUCCESS in 57m 00s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ffcd92b040ca4c74a6eeb712e6de3097 : SUCCESS in 58m 10s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/75e7e9fd7d874e479d4ef44953b57637 : SUCCESS in 1h 02m 05s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/be4000fdbcae4e0aaf9ae647e06055e6 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9bbada9b7a2d874e981d6572b83229cb8d9cd045","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-09 12:18:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebb4b9d338948944984141ca58e4b2b9c3d0f6ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:18:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef2f0e4794f583b43d1717fbbb97ab05b6cad6a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 13:26:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b4cce9e49e3a57bc7d15c0fd211ebd567e55e0b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 13:26:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc6ae46f41340ccb651bfe7df7a9be7 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90b9f1f33fc24fa3b41f9c7a84480bc7 : SUCCESS in 3m 56s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/3f1e75bbf506441a8474bca4f3eef0c9 : SUCCESS in 40m 22s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fd641d8d8e4c4498aecb205a7c3a0fed : SUCCESS in 43m 23s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/852a0fa772a94f9684aeebee0d664a1b : SUCCESS in 50m 51s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8e894137a6ed460ead31e556cc170d0d : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f907e26cc46f5528ee8afb7b48a749797c02287","revisions":{"4f907e26cc46f5528ee8afb7b48a749797c02287":{"kind":"REWORK","_number":1,"created":"2020-01-09 11:06:32.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/11/701711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/11/701711/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/11/701711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/11/701711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/11/701711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/11/701711/1"}}},"commit":{"parents":[{"commit":"53617e23547c6a2987471500ae4666fb57b2642c","subject":"Ensure the Aodh DB is synced","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/53617e23547c6a2987471500ae4666fb57b2642c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-08 11:11:06.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-01-09 11:06:17.000000000","tz":60},"subject":"Pin gettext to \u003c 3.3.0","message":"Pin gettext to \u003c 3.3.0\n\nAlso pinning gettext as version 3.3.0 is not compatible with the ruby\nversion in CentOS7.\n\nChange-Id: I8d226aa2218eafc07119f88eca5be24c3514b717\n(cherry picked from commit 3be5458f4b5a96ddae8d7c1dc733fd5ed3f400d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f907e26cc46f5528ee8afb7b48a749797c02287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f907e26cc46f5528ee8afb7b48a749797c02287"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
