)]}'
{"id":"x%2Fpackstack~706175","triplet_id":"x%2Fpackstack~master~I92a8315421d57c2ad77e7a4a83568d1aa1cc66c1","project":"x/packstack","branch":"master","topic":"rdo-centos8","hashtags":[],"change_id":"I92a8315421d57c2ad77e7a4a83568d1aa1cc66c1","subject":"[DNM] Test CentOS8 CBS builds","status":"ABANDONED","created":"2020-02-06 04:07:25.000000000","updated":"2020-06-05 06:25:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3918b185ef3452dc588832d53091395d392bc8f1","_number":706175,"virtual_id_number":706175,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-18 10:07:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 16:21:56.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-02-18 10:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13218d052430cf9ea19827f59d72bf6646c0cc93","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-06 04:07:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"389412a1f75196ddd52099b7734c4b9f9b39d0b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 05:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3078f565b2945f0964dfe5493ef06ef : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f75c9ddbeae4b9687890b9e5020064b : SUCCESS in 3m 38s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/77dbdb44aaae42f4a001f173019ccd04 : SUCCESS in 52m 14s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/591d0fe6c6394555a13896aa15f2f48d : SUCCESS in 52m 10s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ab38543b3bf144a5b6e7d41aa18cda12 : SUCCESS in 54m 28s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1b547edf450c4b85be84bfec6741af67 : SUCCESS in 56m 30s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cb0e45dd0a6949e69062fbae81d8482d : POST_FAILURE in 9m 21s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ff58b90d369d4c7eb87765df45f390a4 : POST_FAILURE in 9m 00s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/831b31a4207d4912a6865a828a2d3f9b : POST_FAILURE in 10m 11s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d647ee6d6ffe4773a8cde2714280748f : POST_FAILURE in 10m 40s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb5a20042cb431f9f2fc801f34b9a81 : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"761e4b8e453620e464652cef3be7354c60b0e4c7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-07 05:39:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8570904ff5dbd15d41368ac60ba2ec0b34c6ec33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 06:42:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b51557eb625442587bd31ec3695b2c1 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61b9fe79bbce49da9911b5961ec9269f : FAILURE in 2m 47s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/6cc0cc81a07c441ab1748d6d290549fa : SUCCESS in 50m 11s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/24ed72be424943e0a835aa9c587212b9 : SUCCESS in 52m 11s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/5f3d15f251594e379218fca67611a392 : SUCCESS in 50m 43s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/678b3b79a1794415942977795cd60036 : SUCCESS in 51m 49s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3e4b8fc5ca964206983e3b49180e60d0 : SUCCESS in 46m 38s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3e138d319c8f4a15b73a0f472d0789b7 : SUCCESS in 48m 05s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/77764b26e67d48f99ff8b47826d4442c : SUCCESS in 57m 33s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e0ff2e136a9b43daab29ae92b182f840 : SUCCESS in 52m 43s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9478865093934dd881200032dc8c7114 : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03dbcf0432c7f15ca46d3bb90e6db6c1ec561e7b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-17 16:21:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5851efb5fb912aaf5947887cd868430888a8d12f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 17:40:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e627fd0c2f74ebeb1c93d80f71b71a6 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a06804455b2f432fb006ba8b86e0eb9a : SUCCESS in 5m 45s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/9cce66a573f3465cbd3752079c97d3ca : FAILURE in 20m 54s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8a9b494e9c7c48fcbd4b510e740508dd : SUCCESS in 48m 36s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/fa6888da4b0542b593ce7c96d3c49247 : SUCCESS in 55m 17s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1a9c3f1d91564a6f95e13cd9871af829 : SUCCESS in 54m 24s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7e2e8b7ebbdd47f69a78bbf5e5472289 : FAILURE in 17m 24s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c095c8fa3ed7428d8df1ee04d02b578a : SUCCESS in 50m 51s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/93c6e3f3836f4ce0a9d11ba456d8294f : FAILURE in 19m 51s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5bd6d1c7eb434007b0904610d91f4139 : SUCCESS in 57m 32s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a367238b31524bf4b918cb18632ff3f8 : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de69622bffca6c35944482fe6bcd2387e291ed4e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-18 09:03:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"48fa147533b2a368ba69428f0e938e767963c6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 10:07:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6215ba9e7e1749539415e0f5bfb37055 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5257a679bc34ee481007eeec3e00878 : SUCCESS in 5m 00s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8f20d25b679047d99702c8167f2f3c34 : SUCCESS in 50m 19s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/81e1e14e4ea04795b1b6f90a86cc6e3e : SUCCESS in 51m 06s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/c40ba923aeb448789792bb8ce60504b4 : SUCCESS in 51m 21s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f6713a96935b441588caad5e260af23d : SUCCESS in 59m 02s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/04cf4c39401341c0851be207ca00a351 : SUCCESS in 47m 34s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a09fea5554854bb0adc0042b1fbaa363 : SUCCESS in 51m 53s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/66509c54d4924971b66d65e234273a6e : SUCCESS in 56m 57s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/739e1a329b5d49af88fb7633251851f9 : SUCCESS in 56m 30s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55fda7ee946c43a49f04042517a965ce : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9298bd6429be9a24a52b1b018f6ce753516654e4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-05 06:25:32.000000000","message":"Abandoned\n\nwas just for test","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20f9330970fbf20fbd2bc1ace01cfa8514a4e1e9","revisions":{"8bb4de699d0add5445138ff69b59d8993c329aff":{"kind":"REWORK","_number":1,"created":"2020-02-06 04:07:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/75/706175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/706175/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/706175/1"}}},"commit":{"parents":[{"commit":"062b3d0f520cb4f3e5e8e3723f71b18573357f0a","subject":"Use keystone::bootstrap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/062b3d0f520cb4f3e5e8e3723f71b18573357f0a"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-06 04:06:32.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-06 04:06:36.000000000","tz":330},"subject":"[DNM] Test CentOS8 CBS builds","message":"[DNM] Test CentOS8 CBS builds\n\nChange-Id: I92a8315421d57c2ad77e7a4a83568d1aa1cc66c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8bb4de699d0add5445138ff69b59d8993c329aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8bb4de699d0add5445138ff69b59d8993c329aff"}]},"branch":"refs/heads/master"},"20f9330970fbf20fbd2bc1ace01cfa8514a4e1e9":{"kind":"REWORK","_number":2,"created":"2020-02-18 09:03:34.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/75/706175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/706175/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/706175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/706175/2"}}},"commit":{"parents":[{"commit":"062b3d0f520cb4f3e5e8e3723f71b18573357f0a","subject":"Use keystone::bootstrap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/062b3d0f520cb4f3e5e8e3723f71b18573357f0a"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-06 04:06:32.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-18 09:03:31.000000000","tz":0},"subject":"[DNM] Test CentOS8 CBS builds","message":"[DNM] Test CentOS8 CBS builds\n\nChange-Id: I92a8315421d57c2ad77e7a4a83568d1aa1cc66c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20f9330970fbf20fbd2bc1ace01cfa8514a4e1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20f9330970fbf20fbd2bc1ace01cfa8514a4e1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
