)]}'
{"id":"x%2Fpackstack~709966","triplet_id":"x%2Fpackstack~stable%2Fstein~Id6e285e37759aadae9ebf27522d3145b77a6d691","project":"x/packstack","branch":"stable/stein","topic":"swift-logs-stable/train-stable/stein","hashtags":[],"change_id":"Id6e285e37759aadae9ebf27522d3145b77a6d691","subject":"Remove explicit compress of log files","status":"MERGED","created":"2020-02-26 07:55:51.000000000","updated":"2020-02-28 08:12:49.000000000","submitted":"2020-02-28 08:12:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709966-1582877569253-b95df229","meta_rev_id":"69ac73a5cd95b300175da1b723c113c159cf31e9","_number":709966,"virtual_id_number":709966,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-28 06:05:35.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 08:12:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-02-27 13:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 08:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-02-27 13:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 08:55:07.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-02-27 13:25:45.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-02-28 08:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a895d1f22e6153f7ea5bc95281ccec548598c35","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-26 07:55:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"8d23ef200d69be24ef85e478b73b12978d6f4195","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-26 07:56:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 58c8413dd8cf5eeab72c625b1f7faaf781058b88","accounts_in_message":[],"_revision_number":1},{"id":"cba6bb4df78a33fe1f86c0fcf289f9091ffeba53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401ca205630d42748032807d233385c8 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a0dd25197ad47d5ab521a30c9224ccf : SUCCESS in 5m 13s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/ba2999b91537405cabb4d46d64f89fa2 : SUCCESS in 47m 12s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4a8a76036a944d54ab95168e0586a593 : SUCCESS in 39m 04s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/b914fd2052474db1a7f285a4bfaa99c8 : SUCCESS in 55m 42s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/46605cfcb5c84f76ab03ca16a7527153 : SUCCESS in 52m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca9601957b8d4b27949b2ef7c3737f8b : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eab4298ad915b5789ec5822298a748740fa8ac11","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-26 08:55:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"037968ee32b61282b4d5ab33f59841a9584b6b4b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-27 13:25:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b02950bd3c1fee7b15eabb039f9db05a1293657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 13:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b41690b65f840105dd163db0d6af504373de6665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:08:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54ba36508d2471488f4a37f4dc4aa4d : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9bc923e0861424897d5eb1857ded148 : SUCCESS in 4m 29s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/1fb44f18f4714f898a7277c4e63dc151 : SUCCESS in 49m 59s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/af38a39011c64b98a86845c6190f122e : SUCCESS in 42m 57s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/17884d12ef5e4c7f8d7d8d5c8f40232a : SUCCESS in 50m 26s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/762325d170b940f4a4700dc88a1b33de : RETRY_LIMIT in 45m 21s","accounts_in_message":[],"_revision_number":1},{"id":"018420af988e2d3c589d9a87447fea1ccb4ac421","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-28 06:05:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"690e4afbcab39efd497fb285cc36dc9b60efd736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35bc3550e2f945fea5eb0c5f2cdbcb76 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2c230de4e8947a9aa7838ad56c439cd : SUCCESS in 8m 12s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/ad0a85449511483592f206223bd71d17 : SUCCESS in 40m 40s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a666cb1d2c2943ceb88c8ca793eb4beb : SUCCESS in 44m 34s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/73f2226c82664c13bf3d051f996bfc87 : SUCCESS in 52m 25s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b0a231b2ab214eec800517931ae50f43 : SUCCESS in 50m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced63da1ddea45ccb14d39234103bf7d : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ff6097d52e953dc6e5b6adf63bed0eb8dcb2e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:10:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fb6320e42acfc15921dace407379292d7bd03fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:12:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e3da7828c78821dd243e261c616d67fa4491218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:12:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4000dc14044bb5826e6fee0a97a5e7 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/954cbf1ffa6b4967b3628e65e27d36dc : SUCCESS in 7m 11s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/0ab7c70dc7a6489ab9a111f8408f1111 : SUCCESS in 51m 19s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2eb95040daee448abc444a3a2eeff643 : SUCCESS in 58m 01s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/6ee4a013cec4456cb53839b5e553d2c3 : SUCCESS in 45m 21s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2d387fa40dc04e51a0b7749ff14ce3f8 : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00bcd23a2795f0d59cb1113f5c8f100e16224636","revisions":{"00bcd23a2795f0d59cb1113f5c8f100e16224636":{"kind":"REWORK","_number":1,"created":"2020-02-26 07:55:51.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/66/709966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/66/709966/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/66/709966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/66/709966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/66/709966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/66/709966/1"}}},"commit":{"parents":[{"commit":"7cb5aff4b577cbb8c3198b0993ec1da0105c62ca","subject":"Merge \"Ensure the Aodh DB is synced\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7cb5aff4b577cbb8c3198b0993ec1da0105c62ca"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-25 12:11:45.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-02-26 07:55:51.000000000","tz":0},"subject":"Remove explicit compress of log files","message":"Remove explicit compress of log files\n\nExplicit compress is not needed post [1], explicit\ncompress is making log files not browsable.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Id6e285e37759aadae9ebf27522d3145b77a6d691\n(cherry picked from commit 098ae58c6228419280a005073cd41af292b88344)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00bcd23a2795f0d59cb1113f5c8f100e16224636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00bcd23a2795f0d59cb1113f5c8f100e16224636"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
