)]}'
{"id":"x%2Fpackstack~709967","triplet_id":"x%2Fpackstack~stable%2Frocky~Id6e285e37759aadae9ebf27522d3145b77a6d691","project":"x/packstack","branch":"stable/rocky","topic":"swift-logs-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"Id6e285e37759aadae9ebf27522d3145b77a6d691","subject":"Remove explicit compress of log files","status":"MERGED","created":"2020-02-26 07:56:03.000000000","updated":"2020-02-27 14:21:19.000000000","submitted":"2020-02-27 14:21:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709967-1582813279771-577aa814","meta_rev_id":"61abcd7f51bfa96e640211ae7d0a5b32d3796d0f","_number":709967,"virtual_id_number":709967,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 14:21:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-02-27 13:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 08:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-02-27 13:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 08:55:11.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-02-27 13:25:28.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-02-27 14:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29c4e376b3649d9e366b02ce1f31bf68093b8933","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-26 07:56:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"5730bb91033cab333f2b9e0b890308b5c443a28d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-26 07:56:13.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e","accounts_in_message":[],"_revision_number":1},{"id":"45cad180f72bc625f03bfd2c8cac28d406d672db","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-26 08:55:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60ab3e1469a37062e91bbe632f86dee23eacf741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 09:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9241f902cb482bab965c4ce47d22c2 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc9efaa6327c4b3c87fc177a4e798a1e : SUCCESS in 5m 28s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/920d9c63959d454c9f13e553d76038b3 : SUCCESS in 51m 30s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ae603fbba72042af8c95ac9899bbee63 : SUCCESS in 57m 12s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/2adbbbe676aa4e039d8ec1a23b3e8c2f : SUCCESS in 55m 09s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/babe959b852245c294f41324e8e71f04 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0ddfb55b4c5a047e11aeeda6f9d53a1335a62c04","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-27 13:25:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89bd1a42edc45da7d067f365b8727d7739fd3230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 13:25:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e9d3aacf3623e627ba6f2acc922d97e0fcb1a7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:21:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f45654d52c044fe8f17982b71e8ad9a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a30121a82213448b9f0ee373c5575aa0 : SUCCESS in 4m 54s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8efc382c49f34d17ae2e7a007887c33a : SUCCESS in 47m 57s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/955099c2853447f79b923c3106a4c491 : SUCCESS in 47m 33s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/828f672f82a54a299e6ba9a8c73c0bbe : SUCCESS in 48m 24s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/02b4de81f45d415faa0e0ec395ab0673 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a0173d2332a1f87727bf655e55f92189abf4beb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:21:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58c8413dd8cf5eeab72c625b1f7faaf781058b88","revisions":{"58c8413dd8cf5eeab72c625b1f7faaf781058b88":{"kind":"REWORK","_number":1,"created":"2020-02-26 07:56:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/67/709967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/67/709967/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/67/709967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/67/709967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/67/709967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/67/709967/1"}}},"commit":{"parents":[{"commit":"bebff945f334e93726a78ced68cdf48c1458338f","subject":"Pin gettext to \u003c 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bebff945f334e93726a78ced68cdf48c1458338f"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-25 12:11:45.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-02-26 07:56:03.000000000","tz":0},"subject":"Remove explicit compress of log files","message":"Remove explicit compress of log files\n\nExplicit compress is not needed post [1], explicit\ncompress is making log files not browsable.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Id6e285e37759aadae9ebf27522d3145b77a6d691\n(cherry picked from commit 098ae58c6228419280a005073cd41af292b88344)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/58c8413dd8cf5eeab72c625b1f7faaf781058b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/58c8413dd8cf5eeab72c625b1f7faaf781058b88"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
