)]}'
{"id":"x%2Fpackstack~709968","triplet_id":"x%2Fpackstack~stable%2Fqueens~Id6e285e37759aadae9ebf27522d3145b77a6d691","project":"x/packstack","branch":"stable/queens","topic":"swift-logs-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Id6e285e37759aadae9ebf27522d3145b77a6d691","subject":"Remove explicit compress of log files","status":"MERGED","created":"2020-02-26 07:56:13.000000000","updated":"2020-02-28 08:36:50.000000000","submitted":"2020-02-28 08:36:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709968-1582879010711-577aa814","meta_rev_id":"1fefbc815bc626f739a6b0e69c23229c4d2bd49e","_number":709968,"virtual_id_number":709968,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-28 06:05:42.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 08:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-02-27 13:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 08:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-02-27 13:25:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 08:54:54.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-02-27 13:25:53.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-02-28 08:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5919ae9b210f5a47237c35815df63b002ea3bf4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-26 07:56:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"21b555a64f9794283f863bcf1a8005d1f14a2016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:50:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac2d4c4ade048528bcdad19d7130df3 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b2f3fa94fc84a8b81be72168c885727 : SUCCESS in 5m 31s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/433ef8bed2d04ea5a60db748cc8c737b : SUCCESS in 50m 08s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/228c37b33cae48b8a80b1dc5ac7061df : SUCCESS in 47m 14s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/03eae7bdfcef4afe8effbcb38b388a4c : SUCCESS in 51m 57s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c6fd7c7f35744f2fa38ff17bcf9c6ec3 : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0ee5cb3964ca5a07cfdc1867097758e5cdc27c2c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-26 08:54:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9999c45dd6ae88c9a5a76f487ed0cb08d5f1c25","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-27 13:25:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42826f3a670edb0d53536378d3aeda769640d986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 13:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed62b8c179000cdf8af96b034f21db604334d761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:15:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0318a430fc423d8f18bdae3ca92e8c : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c7ca72337ad495f837a356531e0dc10 : SUCCESS in 6m 40s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/e80f72527029473b994722f68124c3ba : RETRY_LIMIT in 37m 28s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/51aa71aff93049d9b0172bb965644753 : RETRY_LIMIT in 30m 48s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ea3cac0cabcc44bb92c73fb4c6e7a1a1 : RETRY_LIMIT in 25m 36s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d5c79288094f4c73ac45bac7d4209d82 : RETRY_LIMIT in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eb31c17583bc5c885ef30aeeca6c386a7f5b3e8c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-28 06:05:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdab05616f2c0ca9e301b83d411eb38ab0365953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a80c2fdcbea4273aaac4ce55552bfd9 : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3aa0750477d4952bd4c033c614fce16 : SUCCESS in 8m 00s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/30b75c95e3db4b3c9c0743fc6842ce5e : SUCCESS in 55m 07s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/956d251d5d5146d1a462071c19feb208 : SUCCESS in 43m 13s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/38a419467a5c4f439ebd866fc100b9ac : SUCCESS in 52m 28s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/498725ea0d5d4b239f6f3869e739b0fa : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8a0fd405cbc921fd88bb18270bdb4e7a3c1ae8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:35:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c13cbb17d2b156b39edf68009c8ebaf7111717fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:36:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea3e21226f048b296b0fc2f9c0ae0be : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13456a4b2d8c436ebe0aac4e000e8b2f : SUCCESS in 6m 37s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/064022ef63cc433bac2749acf5c07bd5 : SUCCESS in 55m 29s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/73c7648a66fd4e0590510266dd66b426 : SUCCESS in 43m 15s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/9f0e3426e38c41c59132e1fe636b406d : SUCCESS in 46m 34s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/65ea42e1d7ac4c7c970d3d6d3725838a : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b5c088805cd363959491593e57000f2a388769f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:36:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e","revisions":{"4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e":{"kind":"REWORK","_number":1,"created":"2020-02-26 07:56:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/68/709968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/68/709968/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/68/709968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/68/709968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/68/709968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/68/709968/1"}}},"commit":{"parents":[{"commit":"4f907e26cc46f5528ee8afb7b48a749797c02287","subject":"Pin gettext to \u003c 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f907e26cc46f5528ee8afb7b48a749797c02287"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-02-25 12:11:45.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-02-26 07:56:13.000000000","tz":0},"subject":"Remove explicit compress of log files","message":"Remove explicit compress of log files\n\nExplicit compress is not needed post [1], explicit\ncompress is making log files not browsable.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Id6e285e37759aadae9ebf27522d3145b77a6d691\n(cherry picked from commit 098ae58c6228419280a005073cd41af292b88344)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
