)]}'
{"id":"x%2Fpackstack~723850","triplet_id":"x%2Fpackstack~master~I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6","subject":"Fix deprecated parameters in Heat and Manila","status":"MERGED","created":"2020-04-28 09:09:02.000000000","updated":"2020-04-28 16:18:40.000000000","submitted":"2020-04-28 16:18:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723850-1588090720850-39316be2","meta_rev_id":"d99f8eb472e6c22c0c96024c701466537fa1aa9f","_number":723850,"virtual_id_number":723850,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 16:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-04-28 15:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-04-28 15:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-04-28 15:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 15:14:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-04-28 15:22:38.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-04-28 16:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01a5c8afd0ec0fde92f5abfd1cf37f144358f683","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-28 09:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f249bc79d1ccbf530c56c78e8dc1d408c862cee3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-04-28 09:18:13.000000000","message":"Patch Set 1:\n\nlet\u0027s also clear:-\n\u001b[1;31mError: Evaluation Error: Error while evaluating a Resource Statement, Class[Manila::Compute::Nova]:\n  has no parameter named \u0027nova_admin_password\u0027\n  has no parameter named \u0027nova_admin_tenant_name\u0027 (file: /var/tmp/packstack/867dced0092f43fab252d9884e85cbeb/modules/packstack/manifests/manila/backend/generic.pp, line: 41, column: 5) on node generic-packstack-001-9230\u001b[0m\n\nand usage of cinder_admin_ in packstack https://review.opendev.org/#/c/723239/4/releasenotes/notes/update-cinder-auth-params-987a691d2e17ac92.yaml\n\nlog:- https://logserver.rdoproject.org/ci.centos.org/weirdo-generic-packstack-scenario001/9230/weirdo-project/logs/latest/manifests/192.168.1.12_controller.pp.log.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"5ae92a5229c7a59a20f91d060e14b04070e802b9","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-28 09:34:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31ca6edf44cf9e1334d59cfea097bc6b748153b4","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-28 12:39:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"294812ac3b37a5db7c94934efda9669648662aaa","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-28 12:57:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a4a59229780fa45421401066bab66c7c27762fb","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-28 13:56:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"993b3df7a35ebc30fd582cfddc9f570b0ff377d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:01:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a537eb9b75ae43c6a9f5420128144071 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6baca9e3cc87444fa869ebdb096aaa85 : SUCCESS in 4m 37s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f1b3992c341a45bfb098ffad699fe325 : SUCCESS in 53m 41s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/eae9bc4a771f4f7d912dc2043806b98a : SUCCESS in 50m 59s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/55bf19b8579743baba5d2886f1e35b71 : SUCCESS in 58m 28s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ba6eaa2d42694d18ac02f62e21933024 : SUCCESS in 56m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee347556870401f8d1fcf5a8334f5f8 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"39c06943c56bcea69e710a3fba6478685a134dad","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-04-28 15:14:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e443f9e5b69e2000b738b928db335af3c165c672","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-04-28 15:22:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"848dcf9924b51a67cd7af938cdf59d26c6553e43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:23:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"40fe8c271ff97ba139ae4cb87c233cc8642f80e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:18:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3345edcb5a244ad39deb4620a559050b : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb8c4500582c4f1c9f8508631f0c1b11 : SUCCESS in 3m 55s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/85cee13e81be45de98c03670e6b209a0 : SUCCESS in 53m 00s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6265ab61dd114979bc083b889a0e1a1a : SUCCESS in 45m 32s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ced2e56066e84fbcb3b21efd34da34c0 : SUCCESS in 48m 18s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c158b5e63fd947a3bbebd18968000f1b : SUCCESS in 52m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc4e88648b3f4b38babdc387a8c3d115 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":4},{"id":"d99f8eb472e6c22c0c96024c701466537fa1aa9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:18:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4f58975d6c396ae7eb0b0764a971e54404d7de79","revisions":{"35e95fc338112cc0e2e640749f747456de217d69":{"kind":"REWORK","_number":1,"created":"2020-04-28 09:09:02.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/50/723850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/50/723850/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/50/723850/1"}}},"commit":{"parents":[{"commit":"27ca9a13f1a89726619fd4423bc991832d15f04c","subject":"Add the packstack-upload-git-mirror job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/27ca9a13f1a89726619fd4423bc991832d15f04c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:08:34.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:08:34.000000000","tz":120},"subject":"Remove heat::engine::heat_watch_server_url","message":"Remove heat::engine::heat_watch_server_url\n\nThis parameter has been removed from puppet-heat in [1].\n\n[1] - https://review.opendev.org/723222\n\nChange-Id: I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/35e95fc338112cc0e2e640749f747456de217d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/35e95fc338112cc0e2e640749f747456de217d69"}]},"branch":"refs/heads/master"},"a83d2eb0b56cde036d80e580585eaafb05baef34":{"kind":"REWORK","_number":2,"created":"2020-04-28 09:34:04.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/50/723850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/50/723850/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/50/723850/2"}}},"commit":{"parents":[{"commit":"27ca9a13f1a89726619fd4423bc991832d15f04c","subject":"Add the packstack-upload-git-mirror job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/27ca9a13f1a89726619fd4423bc991832d15f04c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:08:34.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:32:40.000000000","tz":120},"subject":"Fix deprecated parameters in Heat and Manila","message":"Fix deprecated parameters in Heat and Manila\n\nPuppet-heat removed the heat::engine::heat_watch_server_url parameter\nin [1]. Also, puppet-manila removed deprecated parameters for Nova in\n[2], and deprecated other Cinder parameters in [3].\n\n[1] - https://review.opendev.org/723222\n[2] - https://review.opendev.org/723231\n[3] - https://review.opendev.org/723239\n\nChange-Id: I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a83d2eb0b56cde036d80e580585eaafb05baef34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a83d2eb0b56cde036d80e580585eaafb05baef34"}]},"branch":"refs/heads/master"},"1b7958558658d1386dddd1e1ca651fbab4653524":{"kind":"REWORK","_number":3,"created":"2020-04-28 12:57:28.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/50/723850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/50/723850/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/50/723850/3"}}},"commit":{"parents":[{"commit":"27ca9a13f1a89726619fd4423bc991832d15f04c","subject":"Add the packstack-upload-git-mirror job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/27ca9a13f1a89726619fd4423bc991832d15f04c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:08:34.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 12:56:55.000000000","tz":120},"subject":"Fix deprecated parameters in Heat and Manila","message":"Fix deprecated parameters in Heat and Manila\n\nPuppet-heat removed the heat::engine::heat_watch_server_url parameter\nin [1]. Also, puppet-manila removed deprecated parameters for Nova and\nNeutron in [2], and Cinder parameters in [3].\n\n[1] - https://review.opendev.org/723222\n[2] - https://review.opendev.org/723231\n[3] - https://review.opendev.org/723239\n\nChange-Id: I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1b7958558658d1386dddd1e1ca651fbab4653524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1b7958558658d1386dddd1e1ca651fbab4653524"}]},"branch":"refs/heads/master"},"4f58975d6c396ae7eb0b0764a971e54404d7de79":{"kind":"REWORK","_number":4,"created":"2020-04-28 13:56:01.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/50/723850/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/50/723850/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/50/723850/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/50/723850/4"}}},"commit":{"parents":[{"commit":"27ca9a13f1a89726619fd4423bc991832d15f04c","subject":"Add the packstack-upload-git-mirror job to release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/27ca9a13f1a89726619fd4423bc991832d15f04c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 09:08:34.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-28 13:45:59.000000000","tz":120},"subject":"Fix deprecated parameters in Heat and Manila","message":"Fix deprecated parameters in Heat and Manila\n\nPuppet-heat removed the heat::engine::heat_watch_server_url parameter\nin [1]. Also, puppet-manila removed deprecated parameters for Nova and\nNeutron in [2][3], and Cinder parameters in [4].\n\n[1] - https://review.opendev.org/723222\n[2] - https://review.opendev.org/723231\n[3] - https://review.opendev.org/723232\n[4] - https://review.opendev.org/723239\n\nChange-Id: I41ae3bba8ccb2f3a7fa3e5742e8384e5431b62f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
