)]}'
{"id":"x%2Fpackstack~724632","triplet_id":"x%2Fpackstack~master~Ib58e62768f5b71e68b79002e01f0d779d1c6138b","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Ib58e62768f5b71e68b79002e01f0d779d1c6138b","subject":"Update rake version in puppet module Gemfile, fix linter","status":"MERGED","created":"2020-04-30 10:36:51.000000000","updated":"2020-06-18 08:51:22.000000000","submitted":"2020-06-18 08:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724632-1592470282713-47b710a3","meta_rev_id":"657d559d3fce4ca5c7017e67447a2f0220d374ae","_number":724632,"virtual_id_number":724632,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 08:51:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-06-11 13:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-06-18 07:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 10:37:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-04-30 11:32:18.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-06-18 07:50:33.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-06-18 08:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2358314e5396c20403361bd9db3edc7da1efb894","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-30 10:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0750aebc4a8706e584fb14b76703cf2ad9ca8686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:28:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f429a2ea11643aaaa11216d3fbedd9f : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b96b02c585e24366906991978b8e0cfc : SUCCESS in 3m 45s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f1641f6b2f1e47d4829a73354c1eec0e : FAILURE in 11m 06s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7a90aafea7ea4a5b840ecd9341e2e4ae : SUCCESS in 45m 30s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3fc9e3ea3d894e57a09ae14b40f27e8f : SUCCESS in 49m 39s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3b2d7c54e59a4bc89e68829cb6da53b6 : SUCCESS in 49m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a31c87611de44369f5bf6863c287318 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b8250c3faf96c6bb29cd7f2897e12cf5e9470776","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2020-04-30 11:32:18.000000000","message":"Patch Set 1:\n\nApplying Puppet manifests                         [ \u001b[0;31mERROR\u001b[0m ]\nERROR : Error appeared during Puppet run: 198.72.124.98_controller.pp\nError: Syntax error at \u0027\u003c\u003c\u0027 (file: /var/tmp/packstack/20276a6a3b704a8fa49aa85f257c1ddd/modules/packstack/manifests/manila/backend/generic.pp, line: 25, column: 1) on node centos-8-inapmtl01-0016287364\nYou will find full trace in log /var/tmp/packstack/20200430-104729-9rvz1c0m/manifests/198.72.124.98_controller.pp.log","accounts_in_message":[],"_revision_number":1},{"id":"75e8794ff1f7d30fe9a8367dd38b2d31f7a92f03","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-30 12:52:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7550e46c7ba4ff544dc3aada6dfd0013826a842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 13:53:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e965f77d3064de1ad53d706e8e062a5 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e4defda34744317974bb10c76a8e2fc : SUCCESS in 3m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/34d517aaa3a8428e937d14de55a0a8f6 : SUCCESS in 47m 09s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/40b019f24e374ff5a0e8b6d56aec80c5 : POST_FAILURE in 14m 33s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ac95a8707e414e2596b9d981c6d982a8 : SUCCESS in 51m 52s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/49a0abafe1b8460192c7bacedcb036d0 : SUCCESS in 56m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c5c2b0950f7479599fb703afa278208 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e366f41794d6c8c7620a2c6abf04338bedd76061","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-04-30 14:13:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04071eec45cb1489e77053005eb5ead49f9e0681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:15:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bed77cae0644619963de426177dc123 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56cb77917c28414a9e29f0abb605b8fd : SUCCESS in 5m 19s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/275609c938ae40818678a358489e89ce : SUCCESS in 47m 18s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0d16095d1e624ab5abe93d0f8a24834c : SUCCESS in 54m 57s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b87d78bd446f42dda1a330e0ecb5a6d4 : SUCCESS in 48m 46s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fe6f417a75524c7cb898333bb268ab6e : SUCCESS in 55m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50bae278041d44d1acc5b56b2e1a0724 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"70ae5b833a0e3a05e99b989b201694cda128f814","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-11 10:51:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"632b91e1ee1efb323dc3a6e9aae9a3cb754a332e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1bac280a4440ebb3ac5baba09e6eac : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cca9d1970d84be2bd7c8cf70ea49bec : SUCCESS in 3m 58s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e75d289d71e1496397ae07c3a6abf2a2 : SUCCESS in 43m 46s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e873ae3bee754a159ce21c9d22ffd6b5 : SUCCESS in 41m 53s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3ddac625ce664d87b6f99ad7e84327c3 : SUCCESS in 45m 36s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4160ef79a4484996bd14bb08a434027b : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":3},{"id":"17580cfe92bf0dc4d72c619ea9b8b2af35eb1123","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-11 13:25:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"178f5f1d9e445d4f2624bc55339b289cb42946c5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-18 07:50:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f9c95d84ac5b7602a21e4232b3239de757225ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:50:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"76ee9a426fa905cc4e163adecc20359c09acefc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:51:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"657d559d3fce4ca5c7017e67447a2f0220d374ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:51:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9554b5a8c324d6982cf7636f2cb81e7 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8bb56c30304f65a0110a4ddec3c68a : SUCCESS in 9m 19s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2842f9d1b5424259aac72cadd678621d : SUCCESS in 54m 58s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d5c9256742884d3381f640ab78532644 : SUCCESS in 51m 55s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1276b77f65d6416fb16ffdc635961d3e : SUCCESS in 48m 15s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/52c9a1ce0d1a46ff94475ff6bcb82268 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ad67a6c01b36da0b8d8987b210727f119947686c","revisions":{"19018518b4f99723969fec3e7da51a035a77fb44":{"kind":"REWORK","_number":1,"created":"2020-04-30 10:36:51.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/32/724632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/724632/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/724632/1"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-30 10:34:51.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-30 10:36:49.000000000","tz":120},"subject":"Update rake version in puppet module Gemfile, fix linter","message":"Update rake version in puppet module Gemfile, fix linter\n\nWe needed to update the Rake version required by the Gemfile inside\nthe Packstack Puppet module directory. After that, a rake lint\nexecution showed many linter errors, so fixing them.\n\nChange-Id: Ib58e62768f5b71e68b79002e01f0d779d1c6138b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/19018518b4f99723969fec3e7da51a035a77fb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/19018518b4f99723969fec3e7da51a035a77fb44"}]},"branch":"refs/heads/master"},"41f67c3f9f4223f42650a4834d12e00db27304b7":{"kind":"REWORK","_number":2,"created":"2020-04-30 12:52:58.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/32/724632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/724632/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/724632/2"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-30 10:34:51.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-30 12:52:50.000000000","tz":120},"subject":"Update rake version in puppet module Gemfile, fix linter","message":"Update rake version in puppet module Gemfile, fix linter\n\nWe needed to update the Rake version required by the Gemfile inside\nthe Packstack Puppet module directory. After that, a rake lint\nexecution showed many linter errors, so fixing them.\n\nChange-Id: Ib58e62768f5b71e68b79002e01f0d779d1c6138b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/41f67c3f9f4223f42650a4834d12e00db27304b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/41f67c3f9f4223f42650a4834d12e00db27304b7"}]},"branch":"refs/heads/master"},"ad67a6c01b36da0b8d8987b210727f119947686c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-11 10:51:26.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/32/724632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/32/724632/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/32/724632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/32/724632/3"}}},"commit":{"parents":[{"commit":"b6df8c7ad0ea596b51dc8da814f54f8df5cba64c","subject":"Fix CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b6df8c7ad0ea596b51dc8da814f54f8df5cba64c"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-04-30 10:34:51.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-06-11 10:51:26.000000000","tz":0},"subject":"Update rake version in puppet module Gemfile, fix linter","message":"Update rake version in puppet module Gemfile, fix linter\n\nWe needed to update the Rake version required by the Gemfile inside\nthe Packstack Puppet module directory. After that, a rake lint\nexecution showed many linter errors, so fixing them.\n\nChange-Id: Ib58e62768f5b71e68b79002e01f0d779d1c6138b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ad67a6c01b36da0b8d8987b210727f119947686c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ad67a6c01b36da0b8d8987b210727f119947686c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
