)]}'
{"id":"x%2Fpackstack~725228","triplet_id":"x%2Fpackstack~master~I76dbf86f37a2d2ac12b45d8a455fded5865b33e9","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I76dbf86f37a2d2ac12b45d8a455fded5865b33e9","subject":"Remove usage of deprecated auth_uri in swift proxy","status":"MERGED","created":"2020-05-04 11:01:06.000000000","updated":"2020-05-04 16:29:33.000000000","submitted":"2020-05-04 16:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725228-1588609773775-3c423189","meta_rev_id":"ec08b0733137896a51cd0e6c5d3c72b6926cf94d","_number":725228,"virtual_id_number":725228,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-04 14:12:36.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 16:29:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-04 12:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-04 11:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-05-04 12:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 11:11:30.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-05-04 12:00:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-04 16:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b3e7b1d38afaa3b600ab975eb011efafee46e39","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-04 11:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae4e17c80f4cb4f15dd40b9790dd66a9ec01ad4b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-04 11:11:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7d4232ef2b07b3d8c0ad09fd7631fc1518b3164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:59:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d088655b269c4089ae7fd876f21aba96 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4d2589f3ff842d6b257df8b589a6391 : SUCCESS in 3m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bd5d153dde1a45638ce8dfd705a43b6c : SUCCESS in 44m 00s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4f187e6c84c646d1a2d3d18c9afb41ec : SUCCESS in 42m 51s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ed116e587b2047c8a0748e2ea027c517 : SUCCESS in 46m 28s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0bc6f3b02b1d4002be71e0d567a40f71 : SUCCESS in 53m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea7f99dfc834abb9bc85331b41893bd : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4a918dd9788b3017f9da9fead75f1da2ce32bf93","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-04 12:00:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"218346c118014f8aacdb612bb333a53227fc2a35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 12:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eaaf85d4cb7551058a3b779808eda30825f66b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:06:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004875f02ca34f05bbbb52ab571c4d08 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4243ecba3ef24a32adc1c56a1952fa29 : SUCCESS in 4m 16s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/89be029ce27c410b852882cd861ddb62 : SUCCESS in 51m 57s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dbb275ecb0574ed4a1a90e61d9f36110 : TIMED_OUT in 2h 01m 00s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f0b7165dcf014f96958d00a443031c89 : SUCCESS in 54m 57s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/04192cfc3fb8440590e4086dc72d9500 : SUCCESS in 55m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bda364c90554e92a6054dc00d19f059 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4efd43dab0f45de23f5f6dd075bf6eb8d9ed7d5e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-04 14:12:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6814bdd4f8a580c26446c4380787c20195a2aa71","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-04 15:02:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fcb4eafbb888bbfd87dd8f6713185c52f974e275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:32:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f52d8d07c684493a77417121db99fd5 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/890671d353f742e68e11af6ebb279343 : SUCCESS in 5m 11s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ce705d4aace64d128acc5a0e72d6de37 : SUCCESS in 50m 16s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5bfb8db1087e4227bf47ddbce91bd02e : SUCCESS in 51m 12s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/dee56311658e4b11aa2128eb14234699 : SUCCESS in 1h 01m 40s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/671ad106cd7b4768a474e420d613f069 : SUCCESS in 1h 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cab098f88e14894bd3a2f6e17401c1a : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"824e1fa13ea89a9e43edde5bcb2bd454cbb68562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:32:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7aa6b955be60cbed419e59fd9c09f932d0b9e96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:29:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec08b0733137896a51cd0e6c5d3c72b6926cf94d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fab1b324c1b4bac968cad43f14850e2 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de5a153e461d40beb03f47744219cb30 : SUCCESS in 4m 27s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d3e0ec8492104759b132f758e2fe9ed7 : SUCCESS in 54m 41s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/366678d7064b4d99986136f607a28f5d : SUCCESS in 53m 52s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c31e040ef0034378a3c4e2560d99166d : SUCCESS in 53m 02s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/78c9dbff1d7a4f0b963a7da4dad59631 : SUCCESS in 52m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/389b249dc5fa47a9a35f9d5c43454d57 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cdf7bc511e47c0e3249894d02c0658a9358f720","revisions":{"7cdf7bc511e47c0e3249894d02c0658a9358f720":{"kind":"REWORK","_number":1,"created":"2020-05-04 11:01:06.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/28/725228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/28/725228/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/28/725228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/28/725228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/28/725228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/28/725228/1"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-04 10:55:52.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-04 10:59:17.000000000","tz":330},"subject":"Remove usage of deprecated auth_uri in swift proxy","message":"Remove usage of deprecated auth_uri in swift proxy\n\nauth_uri param is removed in favor of www_authenticate_uri\nin [1], this patch switch to it in swift proxy authtoken.\n\nAlso remove usage of deprecated param \u0027auth_uri\u0027 in\nswift::proxy::ceilometer as its not used[2].\n\n[1] https://review.opendev.org/#/c/725119/\n[2] https://review.opendev.org/#/c/707748/\n\nChange-Id: I76dbf86f37a2d2ac12b45d8a455fded5865b33e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7cdf7bc511e47c0e3249894d02c0658a9358f720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7cdf7bc511e47c0e3249894d02c0658a9358f720"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
