)]}'
{"id":"x%2Fpackstack~726280","triplet_id":"x%2Fpackstack~stable%2Fussuri~I102d7c1b1e7f4a57a88a328dbd7245c8e538df41","project":"x/packstack","branch":"stable/ussuri","hashtags":[],"change_id":"I102d7c1b1e7f4a57a88a328dbd7245c8e538df41","subject":"Revert \"Fix deprecated parameters in Heat and Manila\"","status":"MERGED","created":"2020-05-08 04:16:56.000000000","updated":"2020-05-11 12:53:53.000000000","submitted":"2020-05-11 12:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726280-1589201633469-4f1731f4","meta_rev_id":"f9e7132c00ca1d1bd8cfaffe330196e051246582","_number":726280,"virtual_id_number":726280,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 12:53:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-05-11 11:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 11:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-05-11 11:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 11:25:56.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-05-11 11:34:19.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-11 12:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b76ef7ae140a78e6cc891997e022a7c462a60cf0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-08 04:16:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f8249749403206df014bdc70c1304b4cb48359d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 04:36:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b7d336b0bf46c78e2858ee12bcc400 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a7e1174e0b943a6af3f7c53ebb75924 : SUCCESS in 4m 06s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cd3f3fb4889c4417b49f15c046263fa8 : FAILURE in 14m 21s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/600249dd585e4baea1d84c9034b5cf97 : FAILURE in 14m 46s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/887ed101bb2d4fe782804aac2b7ab596 : FAILURE in 13m 24s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e28421fe626e4fdf8daa2de2fa485a79 : FAILURE in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6eecfee3ad44db6b95141b8eeafda06 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7d100d15eb25bdc5f288096645b26f6c3d2c1995","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-08 05:28:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6aea93e6dfa56a8a0b7aae2e68425a635ce1b3e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 06:25:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6420c32072470e990a9ee48fbae9e3 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9583cad5e404736baecc2258866d46c : SUCCESS in 3m 43s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b0c13b46f585404f9c30d738c134d700 : SUCCESS in 44m 33s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e16f500cbfc64b9ea72a9ba79be11b6e : SUCCESS in 43m 20s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c5d1ec8fe0744242aa18d87eb93b7f40 : SUCCESS in 52m 10s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6dd82741ad17482ebcae2e6b9c6a27ee : SUCCESS in 52m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/097fc07ffad34166b0b2a760534ae1ac : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"045bdddab9546b46c4a03c0cbe7202c6bad0fc49","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-08 10:17:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e03ff40171aebc8d561033f71514b50cd21364c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:24:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad0fd9d59d94487af44be54b078d34b : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1445bc656814c5984b3ce6baed7c981 : SUCCESS in 7m 46s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/061e5e0c1eee4973bb7b0c89b2621441 : SUCCESS in 58m 35s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f1248933fc534a4684d5b4383b3f1b70 : SUCCESS in 50m 34s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7a6750da9a2d433188b0ed844339d353 : SUCCESS in 53m 21s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/56816558297d4798bfc2c15b8f3c846e : SUCCESS in 1h 02m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7f9bfbaf6d445095eb48468430f9fe : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"276dbde6e02231444d0951c97adb0151572044db","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-11 11:25:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ebc7dfe1d6c933fab455c1d9301e30e44df8ee17","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-11 11:34:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6e884f28e853c6e0443ecaf1a7de077bcd5866b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:34:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b69a8f1dcae9c493f8d6a0abe0d18a9c44844ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:53:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f9e7132c00ca1d1bd8cfaffe330196e051246582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:53:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c53ebe8a9e455cb587e786d82c555c : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9876a813294c4271ac446b60cfe9a2ef : SUCCESS in 8m 14s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1798f64c8d29419fb2984348a8e9bdfe : SUCCESS in 50m 48s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4e2c000a11374951a953334176ff0d0e : SUCCESS in 54m 15s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/33a74f1f35fa476e8b911436a37665f6 : SUCCESS in 56m 42s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e70baf06d76447f49e7f5cdfa6e6248b : SUCCESS in 1h 17m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/399209e4b69640679abac33318a0c73d : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4cebf5b2830b8d16595be9066b59a5fbf0fce4ac","revisions":{"c1b5958063a0f084aaedc51a551f77d34883b541":{"kind":"REWORK","_number":1,"created":"2020-05-08 04:16:56.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/80/726280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/726280/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/726280/1"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 03:57:07.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 04:15:22.000000000","tz":330},"subject":"Revert \"Fix deprecated parameters in Heat and Manila\"","message":"Revert \"Fix deprecated parameters in Heat and Manila\"\n\nThe mentioned commits of puppet-heat and puppet-manila\nare not in stable/ussuri so the patch is not needed in\nussuri.\n\nThis reverts commit 4f58975d6c396ae7eb0b0764a971e54404d7de79.\n\nChange-Id: I102d7c1b1e7f4a57a88a328dbd7245c8e538df41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c1b5958063a0f084aaedc51a551f77d34883b541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c1b5958063a0f084aaedc51a551f77d34883b541"}]},"branch":"refs/heads/stable/ussuri"},"86e1c8c8c9149f282d300364d52d8e7b91221941":{"kind":"REWORK","_number":2,"created":"2020-05-08 05:28:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/80/726280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/726280/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/726280/2"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 03:57:07.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 05:25:16.000000000","tz":330},"subject":"Revert \"Fix deprecated parameters in Heat and Manila\"","message":"Revert \"Fix deprecated parameters in Heat and Manila\"\n\nThe mentioned commits of puppet-heat and puppet-manila\nare not in stable/ussuri so the patch is not needed in\nussuri.\n\nThis reverts commit 4f58975d6c396ae7eb0b0764a971e54404d7de79.\n\nAlso Switch to Ussuri branch and repos.\n\nChange-Id: I102d7c1b1e7f4a57a88a328dbd7245c8e538df41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/86e1c8c8c9149f282d300364d52d8e7b91221941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/86e1c8c8c9149f282d300364d52d8e7b91221941"}]},"branch":"refs/heads/stable/ussuri"},"4cebf5b2830b8d16595be9066b59a5fbf0fce4ac":{"kind":"REWORK","_number":3,"created":"2020-05-08 10:17:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/80/726280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/726280/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/726280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/726280/3"}}},"commit":{"parents":[{"commit":"4f58975d6c396ae7eb0b0764a971e54404d7de79","subject":"Fix deprecated parameters in Heat and Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f58975d6c396ae7eb0b0764a971e54404d7de79"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 03:57:07.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-08 10:16:42.000000000","tz":330},"subject":"Revert \"Fix deprecated parameters in Heat and Manila\"","message":"Revert \"Fix deprecated parameters in Heat and Manila\"\n\nThe mentioned commits of puppet-heat and puppet-manila\nare not in stable/ussuri so the patch is not needed in\nussuri, doing a partial revert as 3/4 commits just removed\ndeprecated parameters, [1] introduced new parameters which\nare not available in stable/ussuri.\n\n[1] https://review.opendev.org/#/c/723239/\n\nThis reverts commit 4f58975d6c396ae7eb0b0764a971e54404d7de79.\n\nAlso Switch to Ussuri branch and repos.\n\nChange-Id: I102d7c1b1e7f4a57a88a328dbd7245c8e538df41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4cebf5b2830b8d16595be9066b59a5fbf0fce4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4cebf5b2830b8d16595be9066b59a5fbf0fce4ac"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
