)]}'
{"id":"x%2Fpackstack~731454","triplet_id":"x%2Fpackstack~stable%2Fussuri~I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","project":"x/packstack","branch":"stable/ussuri","hashtags":[],"change_id":"I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","subject":"Fix CI jobs","status":"MERGED","created":"2020-05-28 07:56:23.000000000","updated":"2020-05-28 11:59:41.000000000","submitted":"2020-05-28 11:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731454-1590667181095-b12ba7e5","meta_rev_id":"191e94c51ebec5070fc3cdfc73e551eeea3c4e42","_number":731454,"virtual_id_number":731454,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 11:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-05-28 10:45:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-28 10:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-05-28 10:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 10:45:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 10:58:14.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-28 11:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d62edcfb54ba856e4851538ec5f05db10e011339","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 07:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3da8df273d4c436cf27a9266ef731f1d19962936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 08:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcfeeb2a899f4e2c9c84b88450ac1ece : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8065491ae7c490da5efdad3b04bb598 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fd59e541c824c5b9c3bdd274f9cedcf : SUCCESS in 4m 35s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cb179f60cbfc481ba33e38b518e7788c : SUCCESS in 54m 17s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/451df1860c9142f5904344353d091089 : SUCCESS in 58m 25s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cff7660234e841149fb652e423b5941c : SUCCESS in 49m 09s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/842bdc918a48477bbffa930e1721f2b9 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"19b96d80cd2b20a3edb0bf4390e98adb9a255d7f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 10:45:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba30c99c074f85f8afaa52a91e9451b8747ccc3e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-28 10:58:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f93dd6f3e021086d1047959c2baf2b53e61dddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:58:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae4c9024606770c607487a6a44e1dbef4554ca23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:59:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbf5c4bdf1e42ba87d8692cf3cfb25a : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6cfb94c86ff49bdbbfa1bd32be7e24c : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2355fae1a3a744cb84ab610ec8f80ff4 : SUCCESS in 3m 54s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/57670f053dc84679b7c2face85491cae : SUCCESS in 49m 44s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f538d8381f67459d9f4dcd042d398cc7 : SUCCESS in 45m 47s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2194e6cc878649019255c337608f8e70 : SUCCESS in 49m 21s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2766fa72c3af4a34a28935121aca20e9 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":1},{"id":"191e94c51ebec5070fc3cdfc73e551eeea3c4e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:59:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88f5683a174e07b8f712f618952e003630b3ae5e","revisions":{"88f5683a174e07b8f712f618952e003630b3ae5e":{"kind":"REWORK","_number":1,"created":"2020-05-28 07:56:23.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/54/731454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/54/731454/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/54/731454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/54/731454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/54/731454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/54/731454/1"}}},"commit":{"parents":[{"commit":"4cebf5b2830b8d16595be9066b59a5fbf0fce4ac","subject":"Revert \"Fix deprecated parameters in Heat and Manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4cebf5b2830b8d16595be9066b59a5fbf0fce4ac"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 07:56:21.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n- Update puppetlabs-apache version to 5.4.0\n\n[1] - https://review.opendev.org/730813\n\nCloses-Bug: #1880681\nDepends-On: https://review.opendev.org/730813\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/88f5683a174e07b8f712f618952e003630b3ae5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/88f5683a174e07b8f712f618952e003630b3ae5e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
