)]}'
{"id":"x%2Fpackstack~731455","triplet_id":"x%2Fpackstack~stable%2Ftrain~I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","project":"x/packstack","branch":"stable/train","hashtags":[],"change_id":"I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","subject":"Fix CI jobs","status":"MERGED","created":"2020-05-28 07:58:35.000000000","updated":"2020-05-28 11:59:40.000000000","submitted":"2020-05-28 11:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731455-1590667180211-ad19e745","meta_rev_id":"265a1ff820c9653c147271710e2ae65cccb8febf","_number":731455,"virtual_id_number":731455,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 11:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-05-28 10:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-28 10:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-05-28 10:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 10:46:23.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 10:57:23.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-28 11:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"967beb92906cafff7ff9dd4004dbe480103812f6","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 07:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5320cd9ef5b08d1e5853debada2ed638eb30bbb","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 08:18:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d72e0fd33486e986b0f2f7d5f83beb8bdbaee06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:15:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541060b6ec9246028a3a64878a8789a5 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5b0ab1cf17c4cd58ca68373d9fc3ea6 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9dc1fef2f014738990685e14687e553 : SUCCESS in 4m 46s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/756713a521b74389bf1dca232fc6f2b0 : SUCCESS in 49m 17s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c44bae96b29c4a6586cc9e0d4905104c : SUCCESS in 38m 53s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ec9d91c1d0a148a0a211f427a0579b0c : SUCCESS in 51m 51s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/368b6313318b48b78559e49852e3ffc8 : SUCCESS in 49m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2373438f1f6b4c12b7c3b81c01f1710d : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a171837051600a784a572caf209ffdecb972ee83","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 10:46:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61b1f9eabf6866e81509df7ac0c07cf71c9a49ff","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-28 10:57:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"00f30e0592193d0aabf0a390ecb54c9cbfb28c27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:57:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"059630371d237c48a1953f2f04a71cdd3c5c5708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:59:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5d70cebd6d4ad4b5fe9f1487ca13ae : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f5193654b7b4f7996e090fd7aef5cb5 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57bbb69241f34f929bb49bb0e5c33867 : SUCCESS in 5m 29s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/4d1910491f744f7cab02874ebf45f2dc : SUCCESS in 45m 26s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4cf0a1aaac1a4d59b32ca259a43d3996 : SUCCESS in 43m 41s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/3fb60436649d492d9c31f0f6a962158d : SUCCESS in 51m 42s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/517bd6a0b1d8439b8e86be37d9437b6b : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":2},{"id":"265a1ff820c9653c147271710e2ae65cccb8febf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"100c950bc7838d691841f8d0f2801eef9579ba5e","revisions":{"bd410526ad58a3641764088458a6556aafbe679a":{"kind":"REWORK","_number":1,"created":"2020-05-28 07:58:35.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/55/731455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/55/731455/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/55/731455/1"}}},"commit":{"parents":[{"commit":"6aaf0c07ca7f37550f7d673fdadd9c534ff8e362","subject":"Fix manage_rdo function for CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6aaf0c07ca7f37550f7d673fdadd9c534ff8e362"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 07:58:06.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd410526ad58a3641764088458a6556aafbe679a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd410526ad58a3641764088458a6556aafbe679a"}]},"branch":"refs/heads/stable/train"},"100c950bc7838d691841f8d0f2801eef9579ba5e":{"kind":"REWORK","_number":2,"created":"2020-05-28 08:18:30.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/55/731455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/55/731455/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/55/731455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/55/731455/2"}}},"commit":{"parents":[{"commit":"6aaf0c07ca7f37550f7d673fdadd9c534ff8e362","subject":"Fix manage_rdo function for CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6aaf0c07ca7f37550f7d673fdadd9c534ff8e362"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 08:18:23.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/100c950bc7838d691841f8d0f2801eef9579ba5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/100c950bc7838d691841f8d0f2801eef9579ba5e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
